)]}'
{"id":"openstack%2Fosc-placement~695629","triplet_id":"openstack%2Fosc-placement~master~Ia5b513f27225f5dbb0cac7ff11a757501a1ca54e","project":"openstack/osc-placement","branch":"master","topic":"generation-1.28","hashtags":[],"change_id":"Ia5b513f27225f5dbb0cac7ff11a757501a1ca54e","subject":"Return consumer generation for microversion 1.28","status":"ABANDONED","created":"2019-11-22 06:24:06.000000000","updated":"2020-09-11 06:39:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"675eb3088792e693a0ed9f7cf1daab821d1a4715","_number":695629,"virtual_id_number":695629,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-25 17:18:46.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-22 06:37:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-11-22 14:27:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 06:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-22 14:27:50.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-25 17:18:46.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"0b34a6604caf140e7372732ff905f83ab8556f43","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-22 06:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da3efb725504ea40695ffe51cc1ff2f263f4898d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 06:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0174d1563ce9495f83afe19245dd96ba : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab412983b81149bba46c5e8f4f362710 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0cc2992d6cf44f7bcc970590804faa1 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/226fd09c67314daa95f5f6afb6677764 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec8279a70bb45099b9613d1b9b2852d : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d647471b3b264ee989bf3dfdef369c66 : SUCCESS in 4m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/426250b8b09e42afb9e253d3db333205 : SUCCESS in 4m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c926247c01d547708bb59cedc4b08f78 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4f4c9a27699811f429eb800bb0cc7ee2e40e56c3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-22 14:27:36.000000000","message":"Patch Set 1:\n\nI personally don\u0027t think we need this as it will just clutter up the output for something that the user isn\u0027t going to provide on the command line. As I\u0027ve said elsewhere, generations are for computers to use and keep track of, not humans using a CLI. If a CLI option loses a race because of a generation conflict then the command fails and the user tries again.\n\nThe only case outputting the generation might be warranted is if the user (admin) needs to get that information to debug some issue where they can\u0027t get around the 409 but with the CLIs not taking generation as input I don\u0027t see how that\u0027s really a problem.","accounts_in_message":[],"_revision_number":1},{"id":"c0370652386760e05068113d5763126f5aa1d0e2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-22 14:27:50.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"8e09634a6275e76113e5b6553fbaf6c1e591ba5f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-22 20:15:00.000000000","message":"Patch Set 1:\n\nBTW my -1 is just a soft -1 to register a vote. If other cores want this then I\u0027ll yield. I\u0027m not going to die on a hill over this.","accounts_in_message":[],"_revision_number":1},{"id":"c64413e24395d2676a66facbac125afa99259712","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 17:18:46.000000000","message":"Patch Set 1:\n\nUgh, I hadn\u0027t noticed until this patch that the existing outputs are labeling the resource provider generation column \"generation\", which is blatantly wrong for \u0027allocations\u0027 operations.\n\nAnd if I understand how OSC is versioned, we can\u0027t go back and change that to resource_provider_generation, can we? (We could do it for the table output, but we would have to maintain some kind of aliasing to \u0027generation\u0027 for backward compatibility with existing scripts using -c, right?)\n\nGiven that, I\u0027m torn as to whether adding consumer_generation actually makes that more or less confusing.\n\nI suppose I\u0027ll no-vote for now.\n\n(BTW, to address Matt\u0027s concern, IMO being able to see generations can be helpful for debugging, so I\u0027m not opposed to displaying them in general.)","accounts_in_message":[],"_revision_number":1},{"id":"8bd510d4e98155c3d9faf6c375564bbcee1a61e2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:39:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb2aeea908304a2d2d69fc8f30289fffc855de4e","revisions":{"fb2aeea908304a2d2d69fc8f30289fffc855de4e":{"kind":"REWORK","_number":1,"created":"2019-11-22 06:24:06.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/29/695629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/29/695629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/29/695629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/29/695629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/29/695629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/29/695629/1"}}},"commit":{"parents":[{"commit":"0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d","subject":"Add support for microversion 1.28 in allocation set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-21 15:33:21.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-21 16:11:19.000000000","tz":0},"subject":"Return consumer generation for microversion 1.28","message":"Return consumer generation for microversion 1.28\n\nThis changes the resource provider allocation set/show/unset command\nfor microversion 1.28 to return consumer generations as well as\nresource provider generations.\n\nChange-Id: Ia5b513f27225f5dbb0cac7ff11a757501a1ca54e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fb2aeea908304a2d2d69fc8f30289fffc855de4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fb2aeea908304a2d2d69fc8f30289fffc855de4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
