)]}'
{"id":"openstack%2Fosc-placement~695864","triplet_id":"openstack%2Fosc-placement~master~I8280075a020fb4a10b82b710d5e0ca997a3028fd","project":"openstack/osc-placement","branch":"master","topic":"catch-up-microversion","hashtags":[],"change_id":"I8280075a020fb4a10b82b710d5e0ca997a3028fd","subject":"Improve tests for warning messages","status":"MERGED","created":"2019-11-25 09:38:45.000000000","updated":"2020-01-15 19:47:07.000000000","submitted":"2020-01-15 19:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695864-1579117413606-f3637170","meta_rev_id":"ca3fcb4b25efb82a6af07be442c26897d5cefe55","_number":695864,"virtual_id_number":695864,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 19:43:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2020-01-15 19:25:20.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2020-01-15 19:25:20.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 10:36:17.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-15 19:25:20.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-15 19:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"190103d268bc120c419fab34fbcf0e330e0fad56","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-25 09:38:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a35fde5dadeac5e4dff7d461678578b0cd43144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:49:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4019d5375f4d7195f7afdaa788dfc0 : FAILURE in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1b4008a389f4f96905aa1e376c56daf : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36fc56a836624451b26aaff452980e55 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81e1e2a7643e4470902292f1d9c2eda1 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3635e60f4b3040bc8963370656aa83ab : SUCCESS in 4m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fe79660947c1462f82e404709850a112 : SUCCESS in 3m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58e7f870a296496abb39bb5d9fc96fe1 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4b24e78bca4ddb2f3b9005b37d0da7489b757a12","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-25 11:00:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b6ac5c6aeff9ae683dcc878a464dbe551687f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:07:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e831763b72ae46259689719d09219a03 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3417f2e0f0841b79a743ecd70d3fd23 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a2a460be3ca4f9b9d068bb1fa485212 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e559c8f6e72945eba13e00ac58cee97c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e92e80ff9c34a1190b08ba92be27fa4 : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d1b3bf72b39b44b5b8e1aae6469c13b1 : SUCCESS in 4m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b65d7b36ede44c08ac6d88f235362be3 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"32fb730bfc678f05956cbf1f830b0dd2d03afbb6","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-26 09:13:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5c25b9f2a7b018ac50c7a6766b51f4b4b3fb3033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 09:22:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee1707bfa9b4575a2b8f75d6ef52b59 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7591fed7fbd54205b1f32f555eee6fa0 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6c930ffbe514f93a62e0d0231e50133 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7874da910c014e6289de767a0416549f : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9586b0f3d56349359ca2fa8188bf36c4 : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/814aea4544354a3db5ac32b085e0c7da : SUCCESS in 4m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54ffe41a9d5e47a8bea6e688484e46ff : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"0e960f638f305cae03cc8e22c69dfcc3c8394a5f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 04:25:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"415a86f2f7043ff1cafe3ac468d560f3e9d4112e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 04:31:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6ed79dfce54e5fa14dee7b853c7186 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f46cfbbb63f42dab08d8b74a0281ca1 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b21f2ddbe45e4a7e80b6e62a4baba839 : SUCCESS in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/076207def9dc4035bf9ca370eb46c096 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946e63d8eba747b5a944d2c855694e18 : SUCCESS in 5m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/069b17d869b54547bfda9ea9a52a785b : SUCCESS in 2m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7be42045a4da49c7b4e3d8124b74b96d : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"12557e838fb734b13fcb944b2e62eab082237a05","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 05:34:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f72bdec5f66d364e4a3967ae779858b8b429832d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 05:40:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5532f208fd614cae876e9c315a82abac : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/622c067f8b8344fdb400b353dc65cf4c : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7626ecc64f2411489c0cad097949ae9 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff75d50dc494f69bb85aaebea9e5212 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2625934575414ad3891c2cefc50cd1f0 : SUCCESS in 4m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/90d90ef062ca45938ac01c9e91b1557e : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":5},{"id":"bc8a00d7dd9c09927377619fd71d7a15953518fa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-13 12:14:03.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nI have a naming nit for a better abstraction.","accounts_in_message":[],"_revision_number":5},{"id":"95efc5a51ce07ed6c6bbf3102503dea7fe96e363","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-14 05:10:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc0a64e61ccaa80e111c9dda86043cfcdfecd2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 05:21:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532fd49c8d2348489ecadac66ae4ff69 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c485d456b524658af54eea53302e660 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/117db91cad324c93bc2dd403fada914c : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920b62cd69e54ba59058fdcdfcc27c9e : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f8f0a53e043434396cb9b154714be35 : SUCCESS in 2m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/91345ea6c0e24c3d9ef62975006a8305 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":6},{"id":"559d407710afb857ffa326f2ad630fb9970bb570","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-14 06:09:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"749c03fca4a4ebdf2d4bd7f90fa68cfd041a6330","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-14 10:36:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"654a8ecb73c0d382d0672f5e1dff1e9314f918f9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-14 19:52:13.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nThis looks like a good change. Couple of nonblocking issues inline. I\u0027ll fast approve if you want to fix those up.","accounts_in_message":[],"_revision_number":6},{"id":"c6e6ec8474cec44071da978bd5a6479b5b722e8f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-15 01:16:43.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"80771139e4bd88856b877cd9b4418e8cddbc2393","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-15 01:18:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6d4e6d01b31d782cbcd7cc45f7f4813a7f95e60b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 01:27:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7410ed566cba4f1081c53e313e73a719 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c12a3a23e42948f39596ec646532670e : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16792065b9f44b0988be9f87f58c8252 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3d7f9263cf44bf8fe9b0da46c5a58d : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63caf0d5dfc2474388e2ce5c29f3fc01 : SUCCESS in 4m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/26ccdecadd1a46009b4ca4c7b1dced59 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":7},{"id":"90107c9ee2c3b9bee7955c0ebd9afc056a83d2b7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-15 19:25:20.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nThanks for the updates. Fast approving since gibi was +2 before.","accounts_in_message":[],"_revision_number":7},{"id":"685bee1d844119a5ba90530dc2fc0f6ee7d8add0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 19:25:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4704b83537f64c060b59245dbac8c040bb126683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 19:43:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281d6d59ab4c489c90398147b5c042fd : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4b065c2cac245f48076cb35b0f58a78 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b0aaaf889e34dda9ba10489a5d59298 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e1d27facbd4c4894889e2304e38e5b : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e77b592c7d8471cbb913487884f917a : SUCCESS in 6m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4edead09b41842d4a6b1b10d81182c1f : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":7},{"id":"4b73babffbde260dc4184fad970612cf3a5fbef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 19:43:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"6548752437dcbadb804604e8f78f40f88a7e9b5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 19:47:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f6dc29d1fe42ca84584d880f2bd1e9 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d80651a389072e59cf2aaf80af3209f5674a077b","revisions":{"9233edfb5e5e24a29c524ef177c2c9a0b3ed47af":{"kind":"REWORK","_number":1,"created":"2019-11-25 09:38:45.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/1"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9233edfb5e5e24a29c524ef177c2c9a0b3ed47af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9233edfb5e5e24a29c524ef177c2c9a0b3ed47af"}]},"branch":"refs/heads/master"},"30350a47b93a8397eac29a49aafedac7c2133223":{"kind":"REWORK","_number":2,"created":"2019-11-25 11:00:03.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/2"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 10:01:29.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/30350a47b93a8397eac29a49aafedac7c2133223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/30350a47b93a8397eac29a49aafedac7c2133223"}]},"branch":"refs/heads/master"},"6e87646e81865d970ec5414eb0c4c5d9386ca980":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-26 09:13:42.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/3"}}},"commit":{"parents":[{"commit":"37f521e53cb83d22121804a940aedd97114e82e3","subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 13:35:09.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6e87646e81865d970ec5414eb0c4c5d9386ca980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6e87646e81865d970ec5414eb0c4c5d9386ca980"}]},"branch":"refs/heads/master"},"2a70f069019892984c7fa679c43aedfd37fc506f":{"kind":"REWORK","_number":4,"created":"2019-12-03 04:25:04.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/4"}}},"commit":{"parents":[{"commit":"37f521e53cb83d22121804a940aedd97114e82e3","subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 13:42:01.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2a70f069019892984c7fa679c43aedfd37fc506f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2a70f069019892984c7fa679c43aedfd37fc506f"}]},"branch":"refs/heads/master"},"df07d184ecf8395f2b7f67351b52bf81eb1d4acf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-06 05:34:17.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/5"}}},"commit":{"parents":[{"commit":"e4ceef795aaa1ea292b67991d44ebbb2cb5851e1","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e4ceef795aaa1ea292b67991d44ebbb2cb5851e1"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-26 14:11:42.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df07d184ecf8395f2b7f67351b52bf81eb1d4acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df07d184ecf8395f2b7f67351b52bf81eb1d4acf"}]},"branch":"refs/heads/master"},"0c750f450b3b470494dc53db102082bd987e3bc7":{"kind":"REWORK","_number":6,"created":"2020-01-14 05:10:33.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/6"}}},"commit":{"parents":[{"commit":"e4ceef795aaa1ea292b67991d44ebbb2cb5851e1","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e4ceef795aaa1ea292b67991d44ebbb2cb5851e1"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-09 09:08:43.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c750f450b3b470494dc53db102082bd987e3bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c750f450b3b470494dc53db102082bd987e3bc7"}]},"branch":"refs/heads/master"},"d80651a389072e59cf2aaf80af3209f5674a077b":{"kind":"REWORK","_number":7,"created":"2020-01-15 01:18:08.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/64/695864/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/695864/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/695864/7"}}},"commit":{"parents":[{"commit":"e4ceef795aaa1ea292b67991d44ebbb2cb5851e1","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e4ceef795aaa1ea292b67991d44ebbb2cb5851e1"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-25 02:33:27.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-10 07:12:15.000000000","tz":0},"subject":"Improve tests for warning messages","message":"Improve tests for warning messages\n\nThe functional test for allocation create did test that it displays\nthe warning message if the microversion is not high enough, but didn\u0027t\nvalidate the returned json result.\n\nThis patch changes it to validate the actual result as well as the\nwarning message.\n\nChange-Id: I8280075a020fb4a10b82b710d5e0ca997a3028fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d80651a389072e59cf2aaf80af3209f5674a077b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d80651a389072e59cf2aaf80af3209f5674a077b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
