)]}'
{"id":"openstack%2Fosc-placement~695961","triplet_id":"openstack%2Fosc-placement~master~I444fa2ec69d3fabb588c5baf58e61c4c80517c13","project":"openstack/osc-placement","branch":"master","topic":"story/2006779","hashtags":[],"change_id":"I444fa2ec69d3fabb588c5baf58e61c4c80517c13","subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","status":"MERGED","created":"2019-11-25 17:31:41.000000000","updated":"2019-11-26 00:42:36.000000000","submitted":"2019-11-26 00:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695961-1574728863334-12b15d9b","meta_rev_id":"226547441518cf7b988efce95eaa39abb537c34a","_number":695961,"virtual_id_number":695961,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 00:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-11-25 17:49:12.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-11-26 00:32:20.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-11-26 00:32:20.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 17:49:12.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-26 00:32:20.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-11-26 00:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bda71981ff9a6ad26aa40f7b3ebc65be7a5fb8e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-25 17:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d97f9a37657e60aedb40c00979455795496732c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ebfd55aae940cab689aaa988ac4cd6 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cd68fb5454046f7993da88e056ea124 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4fabf0e623943ca9a4029a6ed4b4bbf : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a562b7aab08b43ffb676de823816980c : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0ff18b4abe40f58405d2613795f646 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cf5879166e34032af7a93119699989f : SUCCESS in 4m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e66ee8035624f1dacc6dabdb1b25d51 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74dbc875e14948fd9364c8981471a9f4 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d6cfe0f1150a710e87407a8db1e1a8944cfa2e12","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 17:49:12.000000000","message":"Patch Set 1: Code-Review+2\n\nGood, thanks Matt.","accounts_in_message":[],"_revision_number":1},{"id":"cf52d2fcf12351c7901c5938c18408888b0183f7","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-26 00:32:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb10e2589def77c1a2b030568d77deb4fe3d79f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:32:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75b6e35e8ebe77707622fc11d1d82b95364c5436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:41:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608c7b84fa8c4d2ba7e7fd5a8a48c2b4 : SUCCESS in 2m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/303fd41f990b4abcbb8ef8cdf9ce3942 : SUCCESS in 2m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e04d4f863eef4eacb3ad4d669f1cf4d1 : SUCCESS in 2m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1441dd6551f34ffb85628131bca8fd55 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd18a176d5f4c5194fcca362a6febd4 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4133e4d35349ed9fd35d15e862f147 : SUCCESS in 2m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2b99085bb1c84e5dba603297f51ab326 : SUCCESS in 3m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca5c65b3aef24d54804dec149a939b6d : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2df5b45027602e00d3caef319ebe960d97485a5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:41:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6bb583391b01d5838db601f07888166b2afb4641","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:42:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b600aeea8dc243afa98c68977ed768a4 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da7bc95508f4ec9a96b688bbc6b1372 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37f521e53cb83d22121804a940aedd97114e82e3","revisions":{"37f521e53cb83d22121804a940aedd97114e82e3":{"kind":"REWORK","_number":1,"created":"2019-11-25 17:31:41.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/61/695961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/61/695961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/61/695961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/61/695961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/61/695961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/61/695961/1"}}},"commit":{"parents":[{"commit":"0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d","subject":"Add support for microversion 1.28 in allocation set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-25 17:30:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-25 17:30:41.000000000","tz":-300},"subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","message":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058\n\nThis addresses comments from that change:\n\n* https://review.opendev.org/#/c/691422/7/osc_placement/resources/allocation.py@207\n* https://review.opendev.org/#/c/691422/7/osc_placement/resources/allocation.py@219\n* https://review.opendev.org/#/c/691422/7/osc_placement/resources/allocation.py@224\n* https://review.opendev.org/#/c/691422/7/releasenotes/notes/allocation-unset-beb0d904c8bc4228.yaml@6\n\nChange-Id: I444fa2ec69d3fabb588c5baf58e61c4c80517c13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
