)]}'
{"id":"openstack%2Fosc-placement~696028","triplet_id":"openstack%2Fosc-placement~master~I2998ff0f3941bf226a942969adf75564a8d5a065","project":"openstack/osc-placement","branch":"master","topic":"support-latest-microversion","hashtags":[],"change_id":"I2998ff0f3941bf226a942969adf75564a8d5a065","subject":"Support auto-negotiated microversion","status":"MERGED","created":"2019-11-26 07:51:57.000000000","updated":"2021-05-12 09:05:02.000000000","submitted":"2021-05-12 09:03:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696028-support-latest-microversion","meta_rev_id":"283710f2d43ba5fc860bc7fb63f6b19c7bf12dc2","_number":696028,"virtual_id_number":696028,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 09:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-12 08:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-05-12 06:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-05-12 08:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 07:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 13:40:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-05-11 11:56:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"391a7de5a59616d5c33a312caf8219ac4303626c","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-26 07:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7960d3448a40f0f1406c0d4a220410092185b90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 08:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1733bc0ce845168ce6f937e48e70bf : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/846163a56cf442c8b70d388e4edee3d1 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d70fde241d64164a0ce3df1aded269d : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73573e8fe4314708a1a986a3572bad2b : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edcb1dac13cb4f2283ee2ea4b4c491a7 : SUCCESS in 5m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ef09c67a76984fd9865a7ee063d95576 : SUCCESS in 3m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48f50e65432f4a88ba63cf16d6cb1591 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a79d849fdb01c402154e6da3d08ae3aa24214b91","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:38:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b1eee2e9b991493732bf84809f45f612f425c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:30:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f79768bde141948aa6b101c3de8961 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cebb1fe5776c4fd5bf54d1383280a98f : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b0e5ae8bb44c669fcfbe2b2b761ee3 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33be0ec57d9340138c76be643050b213 : SUCCESS in 13m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3986ef37dd304a80b47b148180217ca6 : SUCCESS in 4m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/18c404534d0543ddbc5334bdafff7ba7 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e9e9cb3c5b531b3b34aa91be5674dca614429fb6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-28 13:40:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fcdb8471ae1fe1f6dfcaaae8c4f556b5566b3670","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-11 11:56:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fac3beb14e3e320bf57cbb3f1ca0ba2c1d4d4271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:56:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52497273ab9d3f88d45fe0f63b5213053818fe52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 12:06:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d54bf2371f2497890fb5819eb9bd14d : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61e5c9a89c942279ccfb1be48ef9bd9 : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221fe84d67074470b413bcabe1821c37 : FAILURE in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f648a96b3d4457a93e685a20ea61f0 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1985ddd5f8c24de080e805518cf286f6 : SUCCESS in 4m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/60b4ff1060004e06b6e5c29672c48fd1 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b173072a6fa4d82a2e74613e6d7cfc6 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"dcc9522d576afa3c81807e02cddb440da7400512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-12 01:01:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"766e7ce4de525b41232e842447b284566b40459a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:11:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c63c70a3aa84d4cb634892cd8f7271f : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce329609696746f6863cb8541b51dfb6 : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37edc37d79e423183679147bff27ef1 : FAILURE in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bc11d53e3dc49a9aa988549d7012d04 : FAILURE in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f520d6ac2d87442f928d1d8c335cf9e8 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d6ac14ab0b147029f1d415cd036532c : SUCCESS in 4m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c1c6c9cd48d740c0b178975a8763d913 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d00c7080ea914493991557e83c5d5bb1 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ffad7a1de9412c7252114e4b8a0516d3160734d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-12 01:54:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ab4fb7d10528ef8f8da43c08ee3d8a8b6f9e527a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 02:01:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50070c1e595848658be4e492eb828414 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d560d484e9e48cc8640fb20c9028563 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a4ccd57697f402d8734bb4218c4e83a : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04fab5a854e4cbe88678a56eb616f3d : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2808cf7f60fb42e4802a8dc6209be041 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18af61de8b4e424685bde354ff666f12 : SUCCESS in 3m 58s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/dc80801bffaa427db7e9b4a073285ed8 : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0f90246af284f4f963bb5e597689b69 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ced8f4f16d10b3700969c1398a87e71468f7a837","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-12 02:36:04.000000000","message":"Patch Set 4:\n\n```\nThe Captured traceback:\n~~~~~~~~~~~~~~~~~~~\n    Traceback (most recent call last):\n\n      File \"/home/zuul/src/opendev.org/openstack/osc-placement/osc_placement/tests/unit/test_version.py\", line 118, in test_max_version_consistency\n    if versions[j][1] - versions[i][1] !\u003d 1:\n\n    IndexError: list index out of range\n```\n\n^ The gate failure on PS2 is rigid. I\u0027ve just fixed it.","accounts_in_message":[],"_revision_number":4},{"id":"e25f076f2fdef0459c67fd40f4de8574a414401d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-12 06:27:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"557b11116d6699052bfd5374ce064a77d330eebc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-12 08:52:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c287dcb008ab2eb6eb7cc2411caa668738dd9ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:52:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"93e7c807a5755ae0f986a5092143b08201d5c99d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:03:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8c596b99084243b61ab53257681b9d : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa7d68626764411a85808b297e000f6 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37e1abea1d1840c0a04aa53fd759bb84 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72d401b8dd94b21a18f52a4c2e93995 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f058508217645689d7fda755b3f8774 : SUCCESS in 4m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cd4751609e6c41868c6b091a26ff01ac : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a32854c0cf694e91a757d34f80e78724 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"052f0442082dde581fb5fc27a514e177dad6a49a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:03:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"283710f2d43ba5fc860bc7fb63f6b19c7bf12dc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:05:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638c2aa210b14023be8a89bc68798641 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8ac8c8627c6f71af3547799ff25cce098330fb96","revisions":{"b05713d8324ea006833b4a4c1ca903c59e681d5e":{"kind":"REWORK","_number":1,"created":"2019-11-26 07:51:57.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/28/696028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/28/696028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/1"}}},"commit":{"parents":[{"commit":"37f521e53cb83d22121804a940aedd97114e82e3","subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 12:08:41.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 12:08:41.000000000","tz":0},"subject":"Support auto-negotiated microversion","message":"Support auto-negotiated microversion\n\nTo improve the usability for those who are not familiar with\nmicroversion, this patch adds microversion negotiation to pick up max\nmicroversion supported both by client and by server.\n\nThis negotiation is enabled by specifying only major api version,\n``--os-placement-api-version 1``. Note that the default microversion\nremains to be ``1.0``.\n\nChange-Id: I2998ff0f3941bf226a942969adf75564a8d5a065\nStory: #2005448\nTask: #30497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b05713d8324ea006833b4a4c1ca903c59e681d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b05713d8324ea006833b4a4c1ca903c59e681d5e"}]},"branch":"refs/heads/master"},"5f13dc8fdcd8df17669ef65e440d8c94845de9bc":{"kind":"REWORK","_number":2,"created":"2020-09-11 06:38:36.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/28/696028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/28/696028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/2"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 12:08:41.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-09-11 06:32:34.000000000","tz":0},"subject":"Support auto-negotiated microversion","message":"Support auto-negotiated microversion\n\nTo improve the usability for those who are not familiar with\nmicroversion, this patch adds microversion negotiation to pick up max\nmicroversion supported both by client and by server.\n\nThis negotiation is enabled by specifying only major api version,\n``--os-placement-api-version 1``. Note that the default microversion\nremains to be ``1.0`` in this patch and will be switched to the\nnegotiated version in the following patch.\n\nChange-Id: I2998ff0f3941bf226a942969adf75564a8d5a065\nStory: #2005448\nTask: #30497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5f13dc8fdcd8df17669ef65e440d8c94845de9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5f13dc8fdcd8df17669ef65e440d8c94845de9bc"}]},"branch":"refs/heads/master"},"a8026b33cf440dd57821d3f1d68269bbca35bf61":{"kind":"REWORK","_number":3,"created":"2021-05-12 01:01:48.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/28/696028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/28/696028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/3"}}},"commit":{"parents":[{"commit":"ee98b039386febeb9fc14a0f8f064617ff75388c","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ee98b039386febeb9fc14a0f8f064617ff75388c"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 12:08:41.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-12 00:59:59.000000000","tz":0},"subject":"Support auto-negotiated microversion","message":"Support auto-negotiated microversion\n\nTo improve the usability for those who are not familiar with\nmicroversion, this patch adds microversion negotiation to pick up max\nmicroversion supported both by client and by server.\n\nThis negotiation is enabled by specifying only major api version,\n``--os-placement-api-version 1``. Note that the default microversion\nremains to be ``1.0`` in this patch and will be switched to the\nnegotiated version in the following patch.\n\nChange-Id: I2998ff0f3941bf226a942969adf75564a8d5a065\nStory: #2005448\nTask: #30497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a8026b33cf440dd57821d3f1d68269bbca35bf61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a8026b33cf440dd57821d3f1d68269bbca35bf61"}]},"branch":"refs/heads/master"},"8ac8c8627c6f71af3547799ff25cce098330fb96":{"kind":"REWORK","_number":4,"created":"2021-05-12 01:54:16.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/28/696028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/28/696028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/28/696028/4"}}},"commit":{"parents":[{"commit":"ee98b039386febeb9fc14a0f8f064617ff75388c","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ee98b039386febeb9fc14a0f8f064617ff75388c"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-22 12:08:41.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-12 01:52:56.000000000","tz":0},"subject":"Support auto-negotiated microversion","message":"Support auto-negotiated microversion\n\nTo improve the usability for those who are not familiar with\nmicroversion, this patch adds microversion negotiation to pick up max\nmicroversion supported both by client and by server.\n\nThis negotiation is enabled by specifying only major api version,\n``--os-placement-api-version 1``. Note that the default microversion\nremains to be ``1.0`` in this patch and will be switched to the\nnegotiated version in the following patch.\n\nChange-Id: I2998ff0f3941bf226a942969adf75564a8d5a065\nStory: #2005448\nTask: #30497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8ac8c8627c6f71af3547799ff25cce098330fb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8ac8c8627c6f71af3547799ff25cce098330fb96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
