)]}'
{"id":"openstack%2Fosc-placement~696663","triplet_id":"openstack%2Fosc-placement~master~I812e04856df3290a5e4cd7c4ac8f5f47ee566bff","project":"openstack/osc-placement","branch":"master","topic":"microversion-1.30-reshape","hashtags":[],"change_id":"I812e04856df3290a5e4cd7c4ac8f5f47ee566bff","subject":"Support microversion 1.30 - /reshaper API","status":"NEW","created":"2019-11-29 08:20:23.000000000","updated":"2021-05-31 11:05:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3105e637f21007068252645b4e5d1a44c32f3176","_number":696663,"virtual_id_number":696663,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-30 02:31:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-31 11:05:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 07:15:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 14:50:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d85831324be7f6909d1f39228b8484106b7b7f43","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-29 08:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e31180cda29960c433bfd2c7ad0a9e16e0281b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 08:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/94bb67bfeb3845b3880614dcd7d381ef : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2a368072d649c588ec3c2ccf1fd38d : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c9ee4fc10934b67a8411c9a6594be67 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58d116ce11dc4c15b55b03164c4c702a : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a1984b7633a4592893b4118b3d5ec45 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e093230d654ce5b9b542f70171c48a : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/168717f3214a4d3d9a3ead43cfad6afd : SUCCESS in 3m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c9c1676a131144bd8a4fd0cc610aea14 : SUCCESS in 4m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8897ef0ed31494a96150d538dff2040 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1f116281c1fa2174b9c0c62868825a0a11f03390","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-01 13:38:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe1479560c362602f39f8f00b72d40ff1e2d744a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 13:43:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d976cc3b7f44541aa3d6203217ec90f : SUCCESS in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489f4c7c71cc4a9faa25956336e81981 : SUCCESS in 2m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26a2fc5069084693bf3682e91081cd3e : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d5bf5f4cc024b29b97f1424aa6b5202 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0b455bb35974eb4acd170ae2bfb4507 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9950d6b22147a6b811e16a5106714d : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b2baeff09d4feb93236ca2c3e0068b : SUCCESS in 2m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b6c6759f9989438f84c921a458cd9b3d : SUCCESS in 3m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd77db5d003c4aeb81edd36e40fb9c44 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"2d70bdfa1f21df62ef51dcaab5ef948683d8a827","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-02 08:55:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00083cecccb9dff90ce27bc1ff67ce91c55f951b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:03:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff67a8a5de2a4426a16410ebec2889c6 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e9cb3e70eb415f92b4f876eb69e4c4 : SUCCESS in 2m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/800850046ac6463694d6454708a1750f : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcfc4c19acb941eb9503ef897241ae69 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05d6443967894ed2975564200b6d0d2d : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b00970ca5a8481aaabf82c2bff8b072 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/970bc3cac5df4e3486fbebc4f41ecb23 : SUCCESS in 3m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ec629087b04e4301972392670cdc95ec : FAILURE in 4m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2da17f76c92d498bb8af247cfd960189 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3efc2b09c5ca23f48b603568bf78e575724933cb","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 04:25:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f333eac243bcff9db60ea872cd4c63993b126b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 04:32:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fa80bfe28bc494997a9fd771be42f50 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4532857a30eb41a49a99c3d7260b9cf9 : SUCCESS in 2m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d506f8defa444c7b4d68410b81d4aa5 : SUCCESS in 2m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d02f61d9b16e425ead5b8094277cf1e5 : SUCCESS in 2m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/462c571d99594f6b9639cf0380312ec5 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c966144fa64cbca2a0a0ed0e552540 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e75a7a0a7e64919bf07cf9589c9dff3 : SUCCESS in 2m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/db987078c0cb4951ac92e93fcb549595 : FAILURE in 3m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b24287692ccb4c818f2de9afd2baa0af : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"cd1259b431301e41efc6bf71849f1e8e5d58d9a5","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 05:34:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"b466f36b5ee62fa2799326f1517aab8d1cba6c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 05:43:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/031e6d93b2ba406a984fb1873a98d12d : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859b758cf4214f0081478f1d79129ecb : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc6b132496a940fea5ebe35cc1080d6f : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7261f70b66054ccbabc2abbead006d89 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04a1b76cde8423eb885fea25eb342d5 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/016e2c2a595f40179f51db9cf11e3dca : SUCCESS in 2m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/273e6c9829934dd5bf393d54a262262e : SUCCESS in 3m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8a7beb40b1bc437e989be41d5be8873b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":5},{"id":"9a35dccd1c2b0a7da4a98e3c3c318190b0b242be","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-16 00:28:21.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"29172fab7b66d3eb100d0b0940f0fec82428f107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 00:35:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a08b70f8c26f40d199c05d1b168b3bb6 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e461e971f5c4a5ea9cd8e3bed471596 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3769e49eae7441f0b6c919c1fdfeff3f : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9abde7029c44b6db9ae6df6a86135d5 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da4c6576b274bcc8c41fe91e7b45e17 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/236f57c4dade4e39a4e513a6b1877d4d : SUCCESS in 3m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6e177d099c84361b0c0d2ac13574792 : SUCCESS in 2m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/297291e36eb84ec6bc261d32cbfd1bd2 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"7bebedef83eb9cac45755f9c1d271f6a5bb3770f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:24:17.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"467e4dfb5f3ec0c66c64b2dba44cd33fae1aba6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:15:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/697e95e291894d8ebb1849a011baac36 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a434ea86aa3406088ce7a3ad3b9d329 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a875b5bc49543418afd353d062916bc : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f41b0f65cbec4b4dbb9c2fc37fc7bb7f : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273e505b62934430b7438cc969b8d774 : SUCCESS in 15m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6012c0edae496486aa1092648d8342 : SUCCESS in 3m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54299c937bd245ffac78d003edc40741 : FAILURE in 6m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/33eceae57bf04005bd952703f70b0d19 : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":7},{"id":"9db3712ac121324ebe77a85859e34d644375023a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-02 02:02:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"430d963db3744c56b31b76bcf0f6313250a35b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 02:09:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/352de85affca4ffc8097bc0b7e6d13df : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e787f2c1fb4b50bd97e9ab463ef787 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68a6f633530d4a5386b7a07f1769f0c7 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/856f09034b2a4bb386b692259a070f0e : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0be3498f262b417cade0078944b3aaaf : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0dccf53e8d4c2ebea5c4e8172a9108 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63638e6b810e41e6898a5b741da38ce2 : SUCCESS in 3m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1cb6dedae87f4aec94d7b0cc047e3c2b : SUCCESS in 4m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/11f26675324a4679b3e70e38ff900c78 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1902499d822c458aae1e0d1aa485de68 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":8},{"id":"3dc1751747a2b7d90f03a9b43630aa11f07c08e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-13 06:20:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"517c27645028e3e7b99810ef32af82742ddad8cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 06:38:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25ce8db35cbd4fad8a079e4defa80c17 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5734baa43e436dbd0188b316353af4 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ad85a468de342458a3d93fc02b9e253 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77f649fee5714d65bb77eae61a159334 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/453cdfbc6c9f482c89ef7381b2b9ed60 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28098e09e72c4955adfdc39bc7066693 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72608959c960420b902a3b2761639716 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23bce5f343164832a53c92533748f985 : SUCCESS in 6m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7bf3b169bee44d43a766400ecdafbfc8 : SUCCESS in 7m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20072b4388e047b9a5d9c2e4ed72a549 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":9},{"id":"2f2f263d83edd618ab1c39622d430c2bf3069629","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-19 14:50:51.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nQuestion about the abbreviation used in two options. Otherwise, this looks great!","accounts_in_message":[],"_revision_number":9},{"id":"2533157aff933449d00dc72869a8529d2dfea6b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-30 02:17:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"31b4e8b482a8a6fd8533b5c405de81905a6a755a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-30 02:18:25.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8b37827941dae9ee3c37dfc34971634a60683c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-30 02:31:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4e25d94c856d43bf82459796a57d7721 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb640170f3244cb6b73315b8ebaa3c73 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58def6b00dbf42afbcb7920999120975 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04450674f6d24d868650dfff3c3c4de5 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b0a26b2336d4ba195d4dcaa56cef0fc : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfbcd4ede6e408da253818e4423e28c : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d38615655094aa59198bab4fe77af18 : SUCCESS in 3m 57s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ca78bd40720c4c52abb0a80d64860968 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36eeeccf61bc460f890aec575b6e46a9 : SUCCESS in 4m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/45cdd2d55faf4b09bcea25bcc9d466d6 : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d5f5b952d174c1385004a5f0bfed016 : SUCCESS in 4m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2fb54f1dc51047f3a05d901849712e50 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3105e637f21007068252645b4e5d1a44c32f3176","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-31 11:05:17.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"5cf2e20e5f6c43a0af5e1ff33c2e47c118ab49c5","revisions":{"7cab88088f94d94dd0150791a8e25e5255bd0d84":{"kind":"REWORK","_number":1,"created":"2019-11-29 08:20:23.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/1"}}},"commit":{"parents":[{"commit":"37f521e53cb83d22121804a940aedd97114e82e3","subject":"Follow up to I627bfd1ff699d075028da6afafbe7fb9b2f13058","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37f521e53cb83d22121804a940aedd97114e82e3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 11:56:20.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7cab88088f94d94dd0150791a8e25e5255bd0d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7cab88088f94d94dd0150791a8e25e5255bd0d84"}]},"branch":"refs/heads/master"},"c0b95d9cac744c4d89c6fe18d6b0c966f50765f2":{"kind":"REWORK","_number":2,"created":"2019-12-01 13:38:59.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/2"}}},"commit":{"parents":[{"commit":"6e87646e81865d970ec5414eb0c4c5d9386ca980","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6e87646e81865d970ec5414eb0c4c5d9386ca980"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-12-01 13:36:36.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c0b95d9cac744c4d89c6fe18d6b0c966f50765f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c0b95d9cac744c4d89c6fe18d6b0c966f50765f2"}]},"branch":"refs/heads/master"},"c03b354d9feca8bd61bb4fb65642170507a46460":{"kind":"REWORK","_number":3,"created":"2019-12-02 08:55:58.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/3"}}},"commit":{"parents":[{"commit":"6e87646e81865d970ec5414eb0c4c5d9386ca980","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6e87646e81865d970ec5414eb0c4c5d9386ca980"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-12-02 08:55:40.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c03b354d9feca8bd61bb4fb65642170507a46460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c03b354d9feca8bd61bb4fb65642170507a46460"}]},"branch":"refs/heads/master"},"75601f64eed46c8936acfd2ed750b4d978967d49":{"kind":"REWORK","_number":4,"created":"2019-12-03 04:25:04.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/4"}}},"commit":{"parents":[{"commit":"2a70f069019892984c7fa679c43aedfd37fc506f","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2a70f069019892984c7fa679c43aedfd37fc506f"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 14:51:30.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/75601f64eed46c8936acfd2ed750b4d978967d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/75601f64eed46c8936acfd2ed750b4d978967d49"}]},"branch":"refs/heads/master"},"b5c8c454ef1deaaa2baab27e2f27d04b7c367739":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-06 05:34:17.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/5"}}},"commit":{"parents":[{"commit":"df07d184ecf8395f2b7f67351b52bf81eb1d4acf","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df07d184ecf8395f2b7f67351b52bf81eb1d4acf"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-26 14:11:42.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b5c8c454ef1deaaa2baab27e2f27d04b7c367739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b5c8c454ef1deaaa2baab27e2f27d04b7c367739"}]},"branch":"refs/heads/master"},"bb256297a93a1fafd110aacf05ae8132775515a7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-01-16 00:28:21.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/6"}}},"commit":{"parents":[{"commit":"1d6cb4cf0c49728b28191527ded74709128939bc","subject":"Merge \"Improve tests for warning messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1d6cb4cf0c49728b28191527ded74709128939bc"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-11 06:36:36.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bb256297a93a1fafd110aacf05ae8132775515a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bb256297a93a1fafd110aacf05ae8132775515a7"}]},"branch":"refs/heads/master"},"7c96a76c0ee6f9d287d1ae09ed863638d8da2005":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-11 06:24:17.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/7"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-09-11 06:21:22.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7c96a76c0ee6f9d287d1ae09ed863638d8da2005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7c96a76c0ee6f9d287d1ae09ed863638d8da2005"}]},"branch":"refs/heads/master"},"3b27586e19d0ca3ef8a39d2d86bdf5955af25ce2":{"kind":"REWORK","_number":8,"created":"2021-05-02 02:02:10.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/8"}}},"commit":{"parents":[{"commit":"ee98b039386febeb9fc14a0f8f064617ff75388c","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ee98b039386febeb9fc14a0f8f064617ff75388c"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-02 01:39:14.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3b27586e19d0ca3ef8a39d2d86bdf5955af25ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3b27586e19d0ca3ef8a39d2d86bdf5955af25ce2"}]},"branch":"refs/heads/master"},"f3bac06af6ce79118258234250a67d578cd83a67":{"kind":"REWORK","_number":9,"created":"2021-05-13 06:20:40.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/9"}}},"commit":{"parents":[{"commit":"05ca1ee2e16f78524825caf9103205abf6e19497","subject":"Switch default to use latest microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/05ca1ee2e16f78524825caf9103205abf6e19497"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-12 02:04:49.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f3bac06af6ce79118258234250a67d578cd83a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f3bac06af6ce79118258234250a67d578cd83a67"}]},"branch":"refs/heads/master"},"5cf2e20e5f6c43a0af5e1ff33c2e47c118ab49c5":{"kind":"REWORK","_number":10,"created":"2021-05-30 02:17:12.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/63/696663/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/63/696663/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/63/696663/10"}}},"commit":{"parents":[{"commit":"37e47afc88b188a338e3cb70adcf87bac513d4ef","subject":"Merge \"Add openstackclient-plugin-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37e47afc88b188a338e3cb70adcf87bac513d4ef"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 05:17:11.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-30 02:16:41.000000000","tz":0},"subject":"Support microversion 1.30 - /reshaper API","message":"Support microversion 1.30 - /reshaper API\n\nThe new command ``openstack resource provider allocation reshape`` is\nnow supported. This treats an atomic change of a given consumer as\nfollowings.\n\n* Take off the given amount of allocations on the source resource class\n  from the source resource provider\n* Remove the same amount of inventories from the source resource\n  provider\n* Recreate the given amount of inventories of destination resource\n  class in the destination resource provider.\n* Create the same amount of allocations on the destination resource\n  provider for the consumer\n\nAn example to move three VGPU allocations of a consumer from compute RP\nto VGPU RP with those VGPU inventories::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VGPU\u003d3 \\\n    --dst-rp \u003cvgpu_rp_uuid\u003e --add VGPU\u003d3\n\nThe source and destination resource provider can be the same provider.\nAmount can be omitted to move all the allocations for that resource\nclass of the consumer::\n\n  openstack resource provider allocation reshape \u003cconsumer_uuid\u003e \\\n    --src-rp \u003ccompute_rp_uuid\u003e --remove VCPU \\\n    --dst-rp \u003ccompute_rp_uuid\u003e --add PCPU\n\nThis command requires at least ``--os-placement-api-version 1.30``.\n\nChange-Id: I812e04856df3290a5e4cd7c4ac8f5f47ee566bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5cf2e20e5f6c43a0af5e1ff33c2e47c118ab49c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5cf2e20e5f6c43a0af5e1ff33c2e47c118ab49c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
