)]}'
{"id":"openstack%2Fosc-placement~697594","triplet_id":"openstack%2Fosc-placement~master~Ibb679e0cffa18291b91f10b19559d2a68e986486","project":"openstack/osc-placement","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ibb679e0cffa18291b91f10b19559d2a68e986486","subject":"Remove redundant functional-py3* tox environments","status":"MERGED","created":"2019-12-06 00:58:52.000000000","updated":"2019-12-18 16:56:07.000000000","submitted":"2019-12-18 16:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697594-1576688045401-0548d2f5","meta_rev_id":"ed6a8f98ab98de154a2e32c3ded8bc54557bc52e","_number":697594,"virtual_id_number":697594,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 16:54:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-12-18 16:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-06 04:11:10.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-18 16:41:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 04:11:10.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-12-18 16:41:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-18 16:54:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5aa1855bbeb4a80ad9760481070925bc97d4c1dd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-06 00:58:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec5319bcc3318fb1b2e394c86c7337662e74a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a609a1c876c64fcaa7b91417db567ef8 : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4016df234bf6455c99825e3e624015df : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62481d49fbd9432fbfdeaff586c87ef5 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ced912fb0d470e92e2f4f03e49e18c : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f364dfae869d4be29ab899699d83a1a8 : SUCCESS in 3m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e23e3af337244ee9a6d917ffb313271 : SUCCESS in 3m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/751b3b8e85a54572a141c122ece999d6 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9ca4abeb7823b72ee5ead5908ed1f1cfd1526642","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 04:11:10.000000000","message":"Patch Set 1: Code-Review+2\n\nAs long as the Zuul check and gate execute for both environments of py36[1] and py37[2], I am fine with this, thanks. \n\n[1] https://zuul.opendev.org/t/openstack/build/9e23e3af337244ee9a6d917ffb313271/log/job-output.txt#587\n[2] https://zuul.opendev.org/t/openstack/build/751b3b8e85a54572a141c122ece999d6/log/job-output.txt#588","accounts_in_message":[],"_revision_number":1},{"id":"43aee7f6576031678e15f5a90ef7e6392a9d9115","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 16:41:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b1e6c756be3a952b807a8ae9d2f448fa85e53f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:41:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7fa2da9c43afbee91c5bf3829a65dfe831bf30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:54:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e323966ecb4a639c84a68d157cc3d7 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e178ac92797142fb8f37c88a01fbb6b4 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1111c24fddb948cdbdd2dda6cc50481d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21794a7f9ca14660ac332e996aba224e : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc47499cea4c4f6ea79836cf46d049e5 : SUCCESS in 3m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6360455038764e5daefa4fcea2f1a0a3 : SUCCESS in 3m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9b70899054ff4bb3b6589d2c7973150c : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"97450bba2c7eb5cac3e4c520fdaaef7b7d40d5c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:54:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"afe1f9699ae468391d394764b764723b099ee371","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:56:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331c3cd84af54102939e37f1556eb48a : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/919e9dd838d14aa082c319dab27da32f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fb7a255aae7b1a86dd90fdcada59d268997f003","revisions":{"1fb7a255aae7b1a86dd90fdcada59d268997f003":{"kind":"REWORK","_number":1,"created":"2019-12-06 00:58:52.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/94/697594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/94/697594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/94/697594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/94/697594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/94/697594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/94/697594/1"}}},"commit":{"parents":[{"commit":"52ef27a8ce64878c2ef95298e5b9e7370e06f98a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/52ef27a8ce64878c2ef95298e5b9e7370e06f98a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-12-06 00:57:06.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-12-06 00:57:06.000000000","tz":-300},"subject":"Remove redundant functional-py3* tox environments","message":"Remove redundant functional-py3* tox environments\n\nThe functional-py36 and functional-py37 targets are\nnow redundant with their parent functional target\nwhich is now using basepython\u003dpython3 so just running\ntox -efunctional will run with whatever python3 environment\nis on the host, e.g. py36 on standard ubuntu bionic:\n\n  $ python3 --version\n  Python 3.6.8\n\nChange-Id: Ibb679e0cffa18291b91f10b19559d2a68e986486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1fb7a255aae7b1a86dd90fdcada59d268997f003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1fb7a255aae7b1a86dd90fdcada59d268997f003"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
