)]}'
{"id":"openstack%2Fosc-placement~743976","triplet_id":"openstack%2Fosc-placement~master~Iafedac3ff7a5ae60e6027741c44904babf64d3f1","project":"openstack/osc-placement","branch":"master","topic":"usage-in-inventory-list-show","hashtags":[],"change_id":"Iafedac3ff7a5ae60e6027741c44904babf64d3f1","subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","status":"MERGED","created":"2020-07-30 13:50:20.000000000","updated":"2021-03-22 17:34:07.000000000","submitted":"2021-03-22 17:34:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743976-usage-in-inventory-list-show","meta_rev_id":"9ca9ca5d4863f5ddcbc8d6a8aa61c239b96ef300","_number":743976,"virtual_id_number":743976,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-09-11 08:58:05.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 17:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2021-03-22 17:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-09-14 04:23:03.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2021-03-19 15:50:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-22 08:53:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2020-09-04 06:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2021-03-22 17:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 15:58:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-08-06 20:09:27.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-08-07 14:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-04 06:31:32.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-14 04:23:03.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2021-03-19 15:50:15.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"REVIEWER"}],"messages":[{"id":"3f143d826c9ac92679aa5a9e6567848ae84888d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-30 13:50:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff0d1e3381e474979dc342c5a99e1bcb7b9a27c2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-30 13:50:46.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds tests. Waiting for CI so I don\u0027t need to set up a functional env","accounts_in_message":[],"_revision_number":1},{"id":"b33ab78623380c7c92fa92e2dcf7b8387c3b6517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 14:40:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45917410758e4c2684fe0a30a872a10d : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1087c724a5345ef9281d8fa1c676575 : FAILURE in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74f1b4eb55274ceeaa17c1dca5d586a5 : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/110b6b4f900845129369a671d5b7dd0f : FAILURE in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad427c90a274df7b5029580b2fe9bb0 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0efef81e55f4ea6837bd3131ffe1735 : FAILURE in 3m 43s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/297cfbe58d494aa48e59c564bf6c7a91 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a0291628f78ed3962be8350e1e223b4483a3e9d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-05 07:41:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fa18e469ba287989b7e5ae13ef04d95f145d85ac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-05 07:41:25.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9f4124b1cf21c1eae8300ce4c76e4d374bfe9902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 08:02:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0fa0787a324949b09ca6e5eff44ba0 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c30ac7509a8c42598382ce88e4cde0df : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9aed930d20c546e9b956b4d74dc238a6 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9afa38eaf2004e3e976489edc776079d : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa155290ae942d7a888c5df664ba536 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4446bfdbe224ea4b4a2d5f1fdeb459f : SUCCESS in 4m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3acad5bae77b4a07984c3a06bbd16176 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"df5b66be09018738aba00cd4b3149083fc33bfc5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:29:55.000000000","message":"Patch Set 2: -Workflow\n\nwell, wdyk, seems this is perfectly happy","accounts_in_message":[],"_revision_number":2},{"id":"ddcc93d055866ff108934eca3239b2aa52ea7780","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-06 15:58:23.000000000","message":"Patch Set 2: Code-Review-1\n\nSeems useful, and working but needs some test coverage","accounts_in_message":[],"_revision_number":2},{"id":"de94bae2b0da9f11b7ee300eab7eb601124e3e8d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-08-06 20:09:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"850831a1af7911af0d653f52b654d236516a13c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-07 09:38:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3196bb6876fbd62b942f8b48db12217dde1b4b3e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-07 09:39:18.000000000","message":"Patch Set 3:\n\n\u003e Seems useful, and working but needs some test coverage\n\nTurns out we weren\u0027t running functional tests for the past 8 months. Fun! Should be fixed [1] now, though who knows what\u0027s broken in the interim\n\n[1] https://review.opendev.org/#/c/745305/1","accounts_in_message":[],"_revision_number":3},{"id":"287109f0c855513a108ddce97b70120a9c535ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:49:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfda3b01f5846af8881c66e56075998 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763c4be40b3044f0845ca6c38c714f42 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da51cd3e22b40caa17830685c327c1c : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400c5a7fe37441b08bfdb50f52ec32e5 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c08c66c4bb74d708ea6386e0ad2d1a1 : FAILURE in 4m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ee52295d21824a8e970767628f51a96c : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"69497734225d94ac863d9a71423063d0bda10e1f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-07 09:53:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"625dcfc6f1c75d79288067c1a57b746022eeac5d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-07 10:09:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30e28a99a4886e1a25dc6e017fbba35087fdb94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:32:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9216250d76243bdaf8cfe7baabbe1c4 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6669c54d83c94259949dc48a1e5260b6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/982ffd4e4349432f955205bce16e71ef : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4ab1a8ffd04001b819cbb369c31721 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/496ce4dd5d7c478db3f971e91f84a8fb : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48fa22f6e7304bdca05662a9742103dc : FAILURE in 6m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b35a273dca9e45528643d588b3f43c6c : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":4},{"id":"2e17947a8608606419cc8177d16677063a276df4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-07 14:12:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8927a707992ec4c7b1f9eadb01d4dcc8d4e340a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:26:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2779c6eea4eb40368887ce8e27e46176 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2881a01d6c5d4639946ab1e7ea6e4053 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/057bf02aba2f4284bdba992dcadb6f57 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69ed28c5f2b484cbb13509fa12e5816 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72da9dc8dcb64468ad76c45c794caf96 : SUCCESS in 3m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/212996ebd8fe4cc4be17671d8763a4a5 : SUCCESS in 8m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/81d1fa8c93564e44a294f919e2109676 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":5},{"id":"21e0551d37fd643a9d4d51ca40156d1680fea48f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 06:31:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a826869630a7710e045b5a23e38c716946e41d6a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:07:14.000000000","message":"Patch Set 5:\n\nI agree it would be nice to see the usage value within the rp inventory command, and the code looks great, but my (and melanie’s I *guess*?) concern is that if we force this command to fetch usage as well it may degrade the performance in a large scale environment. \n\nIntegrating the usage info to the inventory server API may improve the performance as you wrote in the code comment, but my point is that retrieving the usage costs in the DB layer[1] in the first place compared to retrieving the inventory[2].\n\n[1] https://github.com/openstack/placement/blob/b905ba3ed56652acd9cbff969d76b5363fd5935f/placement/objects/inventory.py#L86\n[2] https://github.com/openstack/placement/blob/master/placement/objects/usage.py#L41\n\n…Any ideas to what extent real operators suffer from such performance degradation?","accounts_in_message":[],"_revision_number":5},{"id":"4fbe4092d5272a76a8d1bbaf5cbd20f3955a9422","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-11 08:43:19.000000000","message":"Patch Set 5:\n\n\u003e I agree it would be nice to see the usage value within the rp\n \u003e inventory command, and the code looks great, but my (and melanie’s\n \u003e I *guess*?) concern is that if we force this command to fetch usage\n \u003e as well it may degrade the performance in a large scale\n \u003e environment.\n \u003e \n \u003e Integrating the usage info to the inventory server API may improve\n \u003e the performance as you wrote in the code comment, but my point is\n \u003e that retrieving the usage costs in the DB layer[1] in the first\n \u003e place compared to retrieving the inventory[2].\n \u003e \n \u003e [1] https://github.com/openstack/placement/blob/b905ba3ed56652acd9cbff969d76b5363fd5935f/placement/objects/inventory.py#L86\n \u003e [2] https://github.com/openstack/placement/blob/master/placement/objects/usage.py#L41\n \u003e \n \u003e …Any ideas to what extent real operators suffer from such\n \u003e performance degradation?\n\nIt\u0027s hard to tell without access to a scale lab or similar, which I don\u0027t have.\n\nI wonder could we ask someone to test this for us? I\u0027ll post to the mailing list and see. We could always add a flag to disable this behavior, however I\u0027d be inclined to add that reactively (i.e. someone tells us there\u0027s a problem) rather than proactively, since no one ever says \"OpenStack\u0027s biggest problem is that it\u0027s not configurable enough\" :P","accounts_in_message":[],"_revision_number":5},{"id":"ccd189c4c46f3f259cdf7fc3dbb18464e8465e38","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-11 08:58:05.000000000","message":"Patch Set 5:\n\n\u003e \u003e I agree it would be nice to see the usage value within the rp\n \u003e \u003e inventory command, and the code looks great, but my (and\n \u003e melanie’s\n \u003e \u003e I *guess*?) concern is that if we force this command to fetch\n \u003e usage\n \u003e \u003e as well it may degrade the performance in a large scale\n \u003e \u003e environment.\n \u003e \u003e\n \u003e \u003e Integrating the usage info to the inventory server API may\n \u003e improve\n \u003e \u003e the performance as you wrote in the code comment, but my point is\n \u003e \u003e that retrieving the usage costs in the DB layer[1] in the first\n \u003e \u003e place compared to retrieving the inventory[2].\n \u003e \u003e\n \u003e \u003e [1] https://github.com/openstack/placement/blob/b905ba3ed56652acd9cbff969d76b5363fd5935f/placement/objects/inventory.py#L86\n \u003e \u003e [2] https://github.com/openstack/placement/blob/master/placement/objects/usage.py#L41\n \u003e \u003e\n \u003e \u003e …Any ideas to what extent real operators suffer from such\n \u003e \u003e performance degradation?\n \u003e \n \u003e It\u0027s hard to tell without access to a scale lab or similar, which I\n \u003e don\u0027t have.\n \u003e \n \u003e I wonder could we ask someone to test this for us? I\u0027ll post to the\n \u003e mailing list and see. We could always add a flag to disable this\n \u003e behavior, however I\u0027d be inclined to add that reactively (i.e.\n \u003e someone tells us there\u0027s a problem) rather than proactively, since\n \u003e no one ever says \"OpenStack\u0027s biggest problem is that it\u0027s not\n \u003e configurable enough\" :P\n\nActually, thinking on this more, I don\u0027t see how this could ever be a problem. The \u0027inventory list\u0027 command lists the inventory for a single resource provider, not multiple resource providers. There\u0027s no reason the response time should get exponentially worse in a larger deployment *unless* placement itself suffers from scale issues (in which case we\u0027re already in trouble, since every API will be impacted). The only significant performance impact we could see if related to the cost of making two round-trip calls to the API. If you wanted, I could spin up a thread pool or something to make these calls in parallel, but that feels like premature optimization and unnecessary complexity","accounts_in_message":[],"_revision_number":5},{"id":"709c9df517c6011ee601ce6b27b46f998875f8c4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-14 04:23:03.000000000","message":"Patch Set 5: Code-Review+2\n\n\u003e I\u0027d be inclined to add flag reactively since no one ever says \"OpenStack\u0027s biggest problem is that it\u0027s not configurable enough\"\n\nRight, much less people say openstack client\u0027s problem is it\u0027s not configurable. \n\n \u003e There\u0027s no reason the response time should get exponentially worse in a larger deployment\n\nWell, I meant there is a use case that operators use the API for every RPs in a larger deployment to retrieve all inventories info in cloud, but... okay, since it can be well multithreaded if they want, I\u0027m sold.","accounts_in_message":[],"_revision_number":5},{"id":"d120a6d39b8ab8dc41bdf9aeb5c641448dd90401","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2021-03-19 15:50:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bd85d13fce658b84945a1eeb41bf724debd7dc7a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-22 08:53:20.000000000","message":"Patch Set 5: Code-Review+1\n\nThis looks good to me but I let Melanie approve it if she agrees.","accounts_in_message":[],"_revision_number":5},{"id":"e7e4c24e6a6ea27efeba0fff9a80f4bbcf7dfc60","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-03-22 17:17:57.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nTo the above discussion, I wasn\u0027t really concerned about performance impact with this proposal but rather wanted to understand more about the reasoning behind adding info from a different API (/usages) to this command (previously only /inventories) given the current state of already having /usages commands. I wanted to get more of an idea of the thoughts around the duplication and get some confidence that this is where \u0027used\u0027 info should be.\n\nI think the explanation on this PS helps give the broader picture of this change and how it relates to what\u0027s already in the CLI. I could see us deprecating \u0027resource provider usage show\u0027 as at least I can\u0027t see what usefulness it has if we have the \u0027used\u0027 info in \u0027inventory list\u0027 and \u0027inventory show\u0027.\n\nThe /resource_providers/usages API has been available since the initial version of the placement API, so it\u0027s safe to include it unconditionally in the \u0027inventory list\u0027 and \u0027inventory show\u0027 commands.\n\nI tried out this change locally in devstack and it looks good.\n\nCode changes LGTM, release note looks good, thanks Stephen!\n\n[1] https://docs.openstack.org/placement/latest/placement-api-microversion-history.html#initial-version","accounts_in_message":[],"_revision_number":5},{"id":"956628cec038529ec9812be193ca3eba94e96779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:18:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"68f8e7defbf3398ad48a8f3a114c24ef8a34f70b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:34:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b591c312e5544e1a1d1f92f94ef97bc : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31ad9f180aac49e9b8b61cef24462be0 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/908f6c8f3a1746758d6392192b46d768 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ea4cb72b8741d7966748d684b8138b : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9801a7991e041a4b5604fd209297121 : SUCCESS in 4m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb42f49ee38240548fea53d15274edb7 : SUCCESS in 6m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6c851b5b2de945f2836b02ce616da95f : SUCCESS in 5m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/413cb4435f2441f791a2b17951f34f64 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"9ca9ca5d4863f5ddcbc8d6a8aa61c239b96ef300","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:34:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dee5bffb286f4bbc403b22d1185dd552a7074545","revisions":{"061a0cd8f946078976a775928b67cfd006eecb56":{"kind":"REWORK","_number":1,"created":"2020-07-30 13:50:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/743976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/76/743976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/1"}}},"commit":{"parents":[{"commit":"bdf3b8c533ab2aaf1c67949bf441a0e6fbad3da3","subject":"trivial: Fix formatting of command help texts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bdf3b8c533ab2aaf1c67949bf441a0e6fbad3da3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-17 15:35:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-30 13:49:55.000000000","tz":60},"subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","message":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027\n\nIt seems odd to show everything about a given resource provider except\nusage. Correct this.\n\nChange-Id: Iafedac3ff7a5ae60e6027741c44904babf64d3f1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/061a0cd8f946078976a775928b67cfd006eecb56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/061a0cd8f946078976a775928b67cfd006eecb56"}]},"branch":"refs/heads/master"},"b16ce1edc6435cb93fa50757c1a1e3801cdd6e5a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-05 07:41:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/743976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/76/743976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/2"}}},"commit":{"parents":[{"commit":"83bf613d38ae6df32bed4b38eb4e1faa95465fa0","subject":"trivial: Fix formatting of command help texts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/83bf613d38ae6df32bed4b38eb4e1faa95465fa0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-17 15:35:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-05 07:41:20.000000000","tz":0},"subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","message":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027\n\nIt seems odd to show everything about a given resource provider except\nusage. Correct this.\n\nChange-Id: Iafedac3ff7a5ae60e6027741c44904babf64d3f1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b16ce1edc6435cb93fa50757c1a1e3801cdd6e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b16ce1edc6435cb93fa50757c1a1e3801cdd6e5a"}]},"branch":"refs/heads/master"},"d04e903cf20cc9e614886de13285a435fb08f84f":{"kind":"REWORK","_number":3,"created":"2020-08-07 09:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/743976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/76/743976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/3"}}},"commit":{"parents":[{"commit":"a7aa5a890cf220ebb6e48716edf02e8311c95b2a","subject":"tox: Trivial cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a7aa5a890cf220ebb6e48716edf02e8311c95b2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-17 15:35:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-07 09:38:10.000000000","tz":60},"subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","message":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027\n\nThe \u0027resource provider inventory list\u0027 and \u0027resource provider inventory\nshow\u0027 commands describe everything about some given inventory *except*\ntheir usage, which is a pretty important metric in quickly assessing\navailability of a given resource. Instead, usage is hidden behind the\n\u0027resource provider usage show\u0027 command. There doesn\u0027t appear to be any\nreason for this to the case, aside from wanting to directly map to the\nplacement API which shouldn\u0027t be a goal. Make these commands more\nfriendly by including usage information. We may want to consider\ndeprecating the \u0027resource provider usage show\u0027 command in the future,\nbut that\u0027s left for another day.\n\nChange-Id: Iafedac3ff7a5ae60e6027741c44904babf64d3f1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d04e903cf20cc9e614886de13285a435fb08f84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d04e903cf20cc9e614886de13285a435fb08f84f"}]},"branch":"refs/heads/master"},"7d17c99e6505cc1bc66bcbb9029e3152f5e71512":{"kind":"REWORK","_number":4,"created":"2020-08-07 10:09:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/743976/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/76/743976/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/4"}}},"commit":{"parents":[{"commit":"a7aa5a890cf220ebb6e48716edf02e8311c95b2a","subject":"tox: Trivial cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a7aa5a890cf220ebb6e48716edf02e8311c95b2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-17 15:35:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-07 10:09:22.000000000","tz":60},"subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","message":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027\n\nThe \u0027resource provider inventory list\u0027 and \u0027resource provider inventory\nshow\u0027 commands describe everything about some given inventory *except*\ntheir usage, which is a pretty important metric in quickly assessing\navailability of a given resource. Instead, usage is hidden behind the\n\u0027resource provider usage show\u0027 command. There doesn\u0027t appear to be any\nreason for this to the case, aside from wanting to directly map to the\nplacement API which shouldn\u0027t be a goal. Make these commands more\nfriendly by including usage information. We may want to consider\ndeprecating the \u0027resource provider usage show\u0027 command in the future,\nbut that\u0027s left for another day.\n\nChange-Id: Iafedac3ff7a5ae60e6027741c44904babf64d3f1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7d17c99e6505cc1bc66bcbb9029e3152f5e71512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7d17c99e6505cc1bc66bcbb9029e3152f5e71512"}]},"branch":"refs/heads/master"},"dee5bffb286f4bbc403b22d1185dd552a7074545":{"kind":"REWORK","_number":5,"created":"2020-08-07 14:12:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/743976/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/76/743976/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/76/743976/5"}}},"commit":{"parents":[{"commit":"a7aa5a890cf220ebb6e48716edf02e8311c95b2a","subject":"tox: Trivial cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a7aa5a890cf220ebb6e48716edf02e8311c95b2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-17 15:35:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-07 14:12:02.000000000","tz":60},"subject":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027","message":"Include usage in \u0027inventory list\u0027, \u0027inventory show\u0027\n\nThe \u0027resource provider inventory list\u0027 and \u0027resource provider inventory\nshow\u0027 commands describe everything about some given inventory *except*\ntheir usage, which is a pretty important metric in quickly assessing\navailability of a given resource. Instead, usage is hidden behind the\n\u0027resource provider usage show\u0027 command. There doesn\u0027t appear to be any\nreason for this to the case, aside from wanting to directly map to the\nplacement API which shouldn\u0027t be a goal. Make these commands more\nfriendly by including usage information. We may want to consider\ndeprecating the \u0027resource provider usage show\u0027 command in the future,\nbut that\u0027s left for another day.\n\nChange-Id: Iafedac3ff7a5ae60e6027741c44904babf64d3f1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/dee5bffb286f4bbc403b22d1185dd552a7074545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/dee5bffb286f4bbc403b22d1185dd552a7074545"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
