)]}'
{"id":"openstack%2Fosc-placement~744358","triplet_id":"openstack%2Fosc-placement~master~I4b622ff3df9dae885fb47bc536a7f5d0e5e05221","project":"openstack/osc-placement","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I4b622ff3df9dae885fb47bc536a7f5d0e5e05221","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-02 00:41:03.000000000","updated":"2020-08-04 22:52:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11bd7bf32f80bc9fc2f2f52d0290b190290acb8d","_number":744358,"virtual_id_number":744358,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 22:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-04 22:25:53.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 22:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e32ad215a45b81743aa63f7939a377e91cd5149","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 00:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3726a7c8e2893a8c7de73c58ca67bcc53ba0793a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 01:12:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d816cf1841450ab35e21950df21ff6 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9095e9de01744e2796773cc7dc524ded : FAILURE in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a294a17f9a8490d8b74b553bcb9a445 : FAILURE in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e8355710f6c4862a7f771cb2c889b34 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0452f6abd4d44ba8cd023fb360ec44a : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae263d016a484ff98433215658af7865 : FAILURE in 2m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/42fc272258c64e51a2b3f2c0e0a69670 : RETRY_LIMIT in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4d1b7b2bca1e9aae57c471ebc99dae7f6bd2853e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 22:47:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"91df8b01dda7c634c332f9b6519a57b4c637e8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:10:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e78939b8a2644aaa9da651b2c61a34d : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c2321e88e484c60b184046f11e9bfae : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13ea6bf155e44b79b047713817322330 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98422def1bd34dcea458b93667258cdf : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb459c5c25640c7b8f5595dc56dc84e : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6553606ba01342b89718710f88c3cc52 : SUCCESS in 3m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2d99ce811fb34e0aba0db585aa75ea91 : RETRY_LIMIT in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e34988708d1b5150c5060224882a317528991aa5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 22:25:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0600fae7f42bf504e9748c7b0eb70dbcdece1161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 22:36:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec5ab3cb02a4e89aae8ef68b6e0e7e8 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9801c87396474fbdb33b6ed919a48f95 : SUCCESS in 2m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f947a3820c14b9fb3d16a70f3187fb4 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9197c7a7086d4a73b66b76994e77c533 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e399788975664ea8994609dca26b0f31 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4546dfd288234cd4a6ec4228f7e9b864 : SUCCESS in 3m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e71e407869ad4984b093fa449823b274 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b8c70554e19785ba0451ff319cb07cdda99a9342","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 22:52:31.000000000","message":"Abandoned\n\nall good here, no updates needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7dc783ddb9fb0c496eddd64855ad2a658ad99ca8","revisions":{"39d42456678bf6311a7720ad36cb0b6237bf9fef":{"kind":"REWORK","_number":1,"created":"2020-08-02 00:41:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/744358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/58/744358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/1"}}},"commit":{"parents":[{"commit":"80bcdd2aaba00888f55294f0b5990bb8569ea015","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/80bcdd2aaba00888f55294f0b5990bb8569ea015"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:39:23.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:39:23.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional tests jobs\n- docs jobs\nDepends-On: https://review.opendev.org/#/c/734700/\nStory: #2007865\nTask: #40209\n\nChange-Id: I4b622ff3df9dae885fb47bc536a7f5d0e5e05221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/39d42456678bf6311a7720ad36cb0b6237bf9fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/39d42456678bf6311a7720ad36cb0b6237bf9fef"}]},"branch":"refs/heads/master"},"7dc783ddb9fb0c496eddd64855ad2a658ad99ca8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-03 22:47:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/744358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/58/744358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/58/744358/2"}}},"commit":{"parents":[{"commit":"df0527de30e40907ace538bc9c75c901470a5466","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df0527de30e40907ace538bc9c75c901470a5466"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 00:39:23.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-03 22:47:25.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional tests jobs\n- docs jobs\nDepends-On: https://review.opendev.org/#/c/734700/\nStory: #2007865\nTask: #40209\n\nChange-Id: I4b622ff3df9dae885fb47bc536a7f5d0e5e05221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7dc783ddb9fb0c496eddd64855ad2a658ad99ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7dc783ddb9fb0c496eddd64855ad2a658ad99ca8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
