)]}'
{"id":"openstack%2Fosc-placement~749855","triplet_id":"openstack%2Fosc-placement~master~I5ec70dc8153dc0ab0bf0323ffb814822566da6cc","project":"openstack/osc-placement","branch":"master","topic":"six_remove","hashtags":[],"change_id":"I5ec70dc8153dc0ab0bf0323ffb814822566da6cc","subject":"Remove six.PY3","status":"MERGED","created":"2020-09-04 01:14:43.000000000","updated":"2020-09-11 01:14:18.000000000","submitted":"2020-09-11 01:13:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749855-1599786787872-b51fe808","meta_rev_id":"64ca5dc46e6f796744bb045dc54b2685bcba54a9","_number":749855,"virtual_id_number":749855,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 01:13:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2020-09-04 15:21:33.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2020-09-11 00:50:24.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-09-11 00:50:24.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 15:21:33.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-09-11 00:50:24.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-09-11 01:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86879332cfcdbf715cb2609051aa85285ffd842c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 01:14:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46dc3fad104db681d5afa1fb3785e2912b6eefdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 01:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a71e57aa5fd49909e7d1160834fbe23 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e662c599997a44ab82e41f2ec803adb0 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af746f7ea4ed45568277b4c1a570ae38 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7c7d946d2d437c8296f7fd557dc8a3 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b49562ed82646338c08267bedac3357 : SUCCESS in 3m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/22bdccf934eb4e1885fb76cee1993dd8 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"851db830b8787a2b3ad07be6e19d134f171751dc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-09-04 15:15:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf9afe291b2b044490904dcec836269af97cec92","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 15:20:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b6505f77f32f6fa45f6a044af7832d391d75165","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-09-04 15:21:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5bf05272da0d7942bed13c8191512ff451fdeed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:29:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42a9b3401144001a18693896bb4af70 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c3f61480f4546dcb699b1092f5bce3d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cec081a716744c8d9e1bf60886aebf21 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440f4262096c4cca947fe6936370587a : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a952f8ec8594f77bb9587e646002c75 : SUCCESS in 3m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4cbd8114500c415d967940e3b307a2e3 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"170283ed11ffe8e245c34e7c76305c43680ffc86","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 00:50:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ded5ed196b769c3dcbd6f7e6b8e00927278a281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 00:52:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04856e75780b87dcfa833cd10a6c25357cc4548a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 01:13:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"500de55903d6a52f09471ab5325fc4249e464b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 01:13:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce70f57e2200417485ff177545cb5e4e : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51dc6733b7a14304b0d5d042e9a79814 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5319f216804bca8a17887c78b14176 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0639754bba48433498d24d606f99a737 : SUCCESS in 16m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91afbd5815714505833b64477c41fa79 : SUCCESS in 3m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6655875783f248d9a9cd6f69a47fe9e4 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2923ae964e8a85492ab8039b3b8c95d5082f6061","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 01:14:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d693db6eaf4140b719ef9df72d2264 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfdebe3c7861d1cd756cbe68458473d288d5dc24","revisions":{"5a144b6175126779c95d78bbf510b2424c645f80":{"kind":"REWORK","_number":1,"created":"2020-09-04 01:14:43.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/55/749855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/55/749855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/1"}}},"commit":{"parents":[{"commit":"df0527de30e40907ace538bc9c75c901470a5466","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df0527de30e40907ace538bc9c75c901470a5466"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 01:13:41.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 01:13:41.000000000","tz":480},"subject":"Remove six.PY3","message":"Remove six.PY3\n\nThe Python 2.7 Support has been dropped since Ussuri.\nSo remove hacking rules for compatibility between python 2 and 3.\n\nChange-Id: I5ec70dc8153dc0ab0bf0323ffb814822566da6cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5a144b6175126779c95d78bbf510b2424c645f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5a144b6175126779c95d78bbf510b2424c645f80"}]},"branch":"refs/heads/master"},"cfdebe3c7861d1cd756cbe68458473d288d5dc24":{"kind":"REWORK","_number":2,"created":"2020-09-04 15:20:33.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/55/749855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/55/749855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/55/749855/2"}}},"commit":{"parents":[{"commit":"df0527de30e40907ace538bc9c75c901470a5466","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df0527de30e40907ace538bc9c75c901470a5466"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 01:13:41.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 15:20:05.000000000","tz":480},"subject":"Remove six.PY3","message":"Remove six.PY3\n\nThe Python 2.7 Support has been dropped since Ussuri.\nSo remove hacking rules for compatibility between python 2 and 3.\n\nChange-Id: I5ec70dc8153dc0ab0bf0323ffb814822566da6cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/cfdebe3c7861d1cd756cbe68458473d288d5dc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/cfdebe3c7861d1cd756cbe68458473d288d5dc24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
