)]}'
{"id":"openstack%2Fosc-placement~756985","triplet_id":"openstack%2Fosc-placement~master~I051511644d75781aa1080fd9f7fadbfc6ce5a902","project":"openstack/osc-placement","branch":"master","topic":"py38","hashtags":[],"change_id":"I051511644d75781aa1080fd9f7fadbfc6ce5a902","subject":"Add py38 package metadata","status":"MERGED","created":"2020-10-09 03:06:44.000000000","updated":"2021-05-14 15:22:05.000000000","submitted":"2021-05-14 15:22:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756985-py38","meta_rev_id":"ca0c47d2c5e7d3539f86224316a6e1e9e63da5d9","_number":756985,"virtual_id_number":756985,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 15:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-14 15:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-05-14 15:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 10:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 15:06:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"bcf780f8bd2dbd9a0eb0cadade4cd700ca8f11fe","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-09 03:06:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd71c7404bf5a8c01a72096dd8635f90de0bd319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:19:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/462086517d1641a7b4b5c3bbeee0ed4c : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4837903931e14ffbb78e50c948c25fa2 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e85071e7d1614e0a963ce3d6c0e731b0 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36a9b6490f314511b3b6984e92e4e014 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eff7096efa34e69aea2f6bdf724226d : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4879c7faaac5425cab5f3ad5f02b7e22 : SUCCESS in 4m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/40479fa5f17e45b8937641eee959cd4d : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c546b2da93d7b5029f3f03ff4445fe8b8962469e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-09 10:19:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2212d220e5fee38316a40b16e5492fa60c15992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 10:29:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a7e6e0bb47640a7beb9c2b0ddff7410 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ece8dd88f83442880784dc5ac2d22bc : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c07e7b75bb44f6da7bf06f89296a38d : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b80060240e641a5ba6ef89c5cb2870a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4c2770b55a4c0bab92904b61400800 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea4c5762249541faba17013e1d18252f : SUCCESS in 4m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2830a8e5a90343978f4dd3a04e236d18 : SUCCESS in 4m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/19603073b6dd461aa83f2fc525ce2a6b : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"11bd82efff31a0a9de74758b6d7f1966ee47acc9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-14 15:06:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50d2ce111f17483e8feb13d335ad3c17174a3569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:07:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a103e1635246039bf24e89ae15f4ca1e128d7bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:22:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b38333f8038340848490d1921e2db7cd : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4297dd0d9f304b32b9bb297d840e4f3d : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c17a9ba5e6dc434ea2c72185f8287f7f : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d82ab2f1b1a34e97a7d4189f7f47c11f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14be7d5b678d494d83aa6466955983c1 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c3e83063d7447197ab68d4ad859660 : SUCCESS in 3m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75d105c8fd6c44f4966080445b63a141 : SUCCESS in 5m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/41b4bc1ccd614fd8b530d21dc87cfb26 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ebacfed906c4a71a177ef2323f4c9b2 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ca0c47d2c5e7d3539f86224316a6e1e9e63da5d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:22:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c16bff5abc398c8172cf918755bf94a4fcf0593b","revisions":{"3f66832b068497940ffb8161565b39c88184b5bd":{"kind":"REWORK","_number":1,"created":"2020-10-09 03:06:44.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/85/756985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/756985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/1"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 03:05:18.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 03:05:18.000000000","tz":480},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nChange-Id: I051511644d75781aa1080fd9f7fadbfc6ce5a902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3f66832b068497940ffb8161565b39c88184b5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3f66832b068497940ffb8161565b39c88184b5bd"}]},"branch":"refs/heads/master"},"c16bff5abc398c8172cf918755bf94a4fcf0593b":{"kind":"REWORK","_number":2,"created":"2020-10-09 10:19:06.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/85/756985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/756985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/756985/2"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 03:05:18.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 10:17:25.000000000","tz":480},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nChange-Id: I051511644d75781aa1080fd9f7fadbfc6ce5a902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c16bff5abc398c8172cf918755bf94a4fcf0593b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c16bff5abc398c8172cf918755bf94a4fcf0593b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
