)]}'
{"id":"openstack%2Fosc-placement~791972","triplet_id":"openstack%2Fosc-placement~master~I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe","project":"openstack/osc-placement","branch":"master","topic":"update-min-tox","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2022-06-06 01:45:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe","subject":"Change minversion of tox to 3.18.0","status":"MERGED","created":"2021-05-18 13:48:34.000000000","updated":"2022-06-06 01:46:35.000000000","submitted":"2022-06-06 01:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791972-update-min-tox","meta_rev_id":"af24656cc2bb2dafe3fb91e1bc5252f7a44419e9","_number":791972,"virtual_id_number":791972,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 01:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-05-31 11:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-05-31 11:31:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 13:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 11:31:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-06-06 01:23:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"375ccfa29109c6d89f262e5cf6f69a54042a548c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 13:48:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8123c1bff71156ad16c342902d38e556e102819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:58:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27df84273ce24bada7905f2766da2998 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac311d3956d64251940650a35a34944e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bf611e407be4fb091666e72a3335051 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e8efc2817af42eda3ba55903a2e4825 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21c0d982c59433eaaf2a0f1acc370a9 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa101818df946c8af1b452edc9baba5 : SUCCESS in 3m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2803e0a317634e11bcf593a59d4966c7 : SUCCESS in 4m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/534fb5c71c5f49a5a491cc336c5526ac : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0cdc383cd07e46f1a2024eea55609cc2 : SUCCESS in 5m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0df97b9f6a0241d8b7f183d57500dec2 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54bcbb9b9f88c9ffd82fdd7fa9e8cf9d5a5e1ae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 14:01:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9dab54bc40d4477704873f6f69c9fbaacc9a444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:09:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0abe646f7a47099231139b14bb395c : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3f6e7d0b444ad5b37828d7663aa0a6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/616dcdd479f748bd85f3934090c4c25d : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c0e2a16409a400e8f719f839f88752c : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6465108ea62b433690257df166bf3959 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75d4ac3d128b49449d0c63528c200974 : SUCCESS in 4m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/910546cdc79d40258db318b012a50e4b : SUCCESS in 6m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/23fadca734904a31a8ae19024b741df9 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b4b9134cb34466c962a72b3153fcd19 : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f726e31d62b14bbc887016e1404a8b21 : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b73be5a3aa76030e5dc880bf13ba78f83f3f369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-06-06 16:13:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"138c9d9cff3e6a431f9b9070b7a7626ea97e8dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-06 16:21:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9311cba21cb14f7dbd5e92c0ab2efc2a : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73106de387e54bbc92182f79eb612e9f : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7622a81ebf942f79e7758ecb8e25405 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43da524532c544d6a292999cc87ff8e3 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66aaad86b783447e8cf8e21edbe02acc : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1849b841d991461b8122685dcdbdc536 : SUCCESS in 4m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b65b8cb6e0464eadadeb4cef55a7bcfc : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/893c75e197f5460cb7aea78727aef102 : SUCCESS in 4m 58s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ef0130c3d40c49b8a8c0a57162053182 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ccf8c9b73de24f8b8b0b5693ff497a45 : SUCCESS in 5m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/026fa43557194a9e803751790e56bb82 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee3f5fdd93ede2ddb276f32730076acf125ede71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-09-04 02:15:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0965659dae7d1279ba3b889bc302307c5f61936a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 02:30:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e819b63c304cc3a9c6936a29e46578 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f491aaa67194df2b14e1d8b77c77dfc : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ba86da011874f23ba9e052a6de0cc8b : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dea791ccf6144b5cacf596f0c398d01b : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815295ef8f8b458f9df9f3297a1ed79b : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6c3ebf39b214daa90bc59baa5aaa9c5 : SUCCESS in 3m 55s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/8cf5b6f9a8ee49819fd961978406f877 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a0fc7d38592f44a39e767057baabd637 : SUCCESS in 5m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8654b0eb779c4a9c8b59e7fcbcf882d6 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/405f248b34944ea3ae04210c39b16a2d : SUCCESS in 4m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2fe1e9780aa54e0e951ff21117d623d8 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c9c8f9ab871286c8b36fbde124e5550c4075b579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-03-29 02:36:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3835df1ec1a52ff217f57bab6acfc3ea6ad2a864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 02:53:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7222e99adc54774b2baa12acebe3db8 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4235229895704a1ab4da9db9af6cbefc : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5253600bd45f4bceab9adc993db07e60 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a039983fedcd4051a285f24ce342a158 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cd1bee0ed5947a7a2640549b4e342ca : SUCCESS in 6m 23s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ba2895b61bc14620bd634000620655bf : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac476f6651ea4a54a5f8c946c5bfec5f : SUCCESS in 7m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e722acae2735487182dece0e9f8a1344 : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77cd2a6732494c519aa117924594a8ab : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/025042a56f634576a60eb663ecc754b5 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":5},{"id":"8fb37f50d6f6ac2e171ba23e21b9cbd1728c4a9a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-31 11:31:20.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aea1872d2bb86fa7047540e56f0d9d1365d42020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:32:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"aecdd6578a8b17f54a952d68b5f9a7687abcaac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:38:21.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9778bcd299204e8295549f5916e9230e : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156032974e2849e59382f16a6d4ca951 : FAILURE in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc98ccf7ad804a81a51309cb119efb47 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2e952118f943588ae8963c2d831a7a : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8c8b337d4c34ea0b82672c8c24bb1f4 : SUCCESS in 3m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cfd6d8b8840245a9842d0dc19fdc7591 : FAILURE in 2m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/64ecfac104c24c1e96d1b9b22e3d1ba8 : FAILURE in 2m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d79d10b93ee24e1b8460d3c75e303e02 : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45dd8886571641088c47a0c1fbf27402 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":5},{"id":"2eb304391b79ef1691f406c3f19a90e1d725097c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-06 01:23:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0da107f855a68f61d2ecb6a278fa15f1d4126b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 01:36:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fae8ae0c243465faf9996fb1e9b7284 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ad7e54768d34bfc89ed02cf828758ef : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb55fc817c174fbbbb4cd558c3613f2b : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6fc0b88aeb7451a821f03e9e93b5cb5 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb55217bbd84a75b0a7d5520e94a1f1 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62490b33b9bf4758afcc2e7672dd1931 : SUCCESS in 2m 56s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5a2acfa35db04d7998e7b39e731aa9aa : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb86f2ee5f5743608544c913b5ec2632 : SUCCESS in 4m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9c05a6916f2479d97db552b92b3bb8b : SUCCESS in 4m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d7413769a2f4921b045a330851383d9 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c9320fbecfb03197608e509b771e35f42586a0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 01:36:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f3210bcbc1ae14f01b8718219619c123ba466266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 01:45:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5b91398c884f2592564f5f3399106f : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53c7fcd93f7e4b019a88b66e37561499 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c191410e918e4a3ab12b80a77f32f80b : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8240044a4447d9a766f2c4c80f7bbf : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf8dc8869db2415885a0ba04df77f6ff : SUCCESS in 2m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a342331751b4cca9c39781952d9b4e4 : SUCCESS in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/20b59f0fbf424e89ada55936e15f9dc9 : SUCCESS in 3m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10444ef290ce4dd3949c8265cea96b37 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2cd649cf6f76971485a8c896f186860a99085b2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 01:45:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"af24656cc2bb2dafe3fb91e1bc5252f7a44419e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 01:46:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f0b5604ae94b6c91ab8b8ecc955b84 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5908c37d1a7448ba8371da9f18cbeb3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d4ef2a949472b9b96a705302cdadcfa6494af07c","revisions":{"e8d24ab5d2cfa047da6db6caf0bdf20adbbd2c82":{"kind":"REWORK","_number":1,"created":"2021-05-18 13:48:34.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/72/791972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/72/791972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/1"}}},"commit":{"parents":[{"commit":"869769e3bfd92ed81ae91290836181515a16875f","subject":"Merge \"Add \"--resource-class\" to  allocation unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/869769e3bfd92ed81ae91290836181515a16875f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:02.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:23.000000000","tz":540},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 python in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e8d24ab5d2cfa047da6db6caf0bdf20adbbd2c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e8d24ab5d2cfa047da6db6caf0bdf20adbbd2c82"}]},"branch":"refs/heads/master"},"ac34c614019f79b157d499b5999aa5c7c0150ac9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-18 14:01:35.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/72/791972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/72/791972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/2"}}},"commit":{"parents":[{"commit":"869769e3bfd92ed81ae91290836181515a16875f","subject":"Merge \"Add \"--resource-class\" to  allocation unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/869769e3bfd92ed81ae91290836181515a16875f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:02.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 14:01:07.000000000","tz":540},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ac34c614019f79b157d499b5999aa5c7c0150ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ac34c614019f79b157d499b5999aa5c7c0150ac9"}]},"branch":"refs/heads/master"},"e158f6e3d02fd758b78432ce77b4cb986738ae8b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-06 16:13:23.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/72/791972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/72/791972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/3"}}},"commit":{"parents":[{"commit":"37e47afc88b188a338e3cb70adcf87bac513d4ef","subject":"Merge \"Add openstackclient-plugin-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/37e47afc88b188a338e3cb70adcf87bac513d4ef"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:02.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-06-06 16:13:23.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e158f6e3d02fd758b78432ce77b4cb986738ae8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e158f6e3d02fd758b78432ce77b4cb986738ae8b"}]},"branch":"refs/heads/master","description":"Rebase"},"4676a2a6b625882b90d48dbdebac8937e5d368f6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-04 02:15:38.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/72/791972/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/72/791972/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/4"}}},"commit":{"parents":[{"commit":"5e38a59317d6941ea7d44446bab1735d719b2564","subject":"Add support for microversion 1.38 consumer types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5e38a59317d6941ea7d44446bab1735d719b2564"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:02.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-09-04 02:15:38.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4676a2a6b625882b90d48dbdebac8937e5d368f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4676a2a6b625882b90d48dbdebac8937e5d368f6"}]},"branch":"refs/heads/master","description":"Rebase"},"d4ef2a949472b9b96a705302cdadcfa6494af07c":{"kind":"REWORK","_number":5,"created":"2022-03-29 02:36:45.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/72/791972/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/72/791972/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/72/791972/5"}}},"commit":{"parents":[{"commit":"7f77e98abd4387f975ba7bf9526043a8b89068d6","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7f77e98abd4387f975ba7bf9526043a8b89068d6"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:36:02.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-03-29 02:36:04.000000000","tz":540},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I004d9563d006a1cce1e2fe7fbcd11fbf41e023fe\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d4ef2a949472b9b96a705302cdadcfa6494af07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d4ef2a949472b9b96a705302cdadcfa6494af07c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
