)]}'
{"id":"openstack%2Fosc-placement~804164","triplet_id":"openstack%2Fosc-placement~stable%2Ftrain~Iad530edaec39545a5573196f4c33ca1b29808fd7","project":"openstack/osc-placement","branch":"stable/train","topic":"update_help-stable/train","hashtags":[],"change_id":"Iad530edaec39545a5573196f4c33ca1b29808fd7","subject":"Update command help information","status":"ABANDONED","created":"2021-08-11 11:21:03.000000000","updated":"2021-09-08 10:29:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":744651,"cherry_pick_of_patch_set":2,"meta_rev_id":"e5df94e52807b2e7cb88f6610a292d88d724689a","_number":804164,"virtual_id_number":804164,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-11 11:28:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 11:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76ce7758c668b5ad007630290b20397bae095473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-11 11:21:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5dd42ebae518ceee0648e3fbcc90a89b0a2cf798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:28:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93e208199a3403bb2b7e80688ef4d6c : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a824355d93a04c2cb53c5276521c145e : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f7855058ddb43e2a19552c5f2a2feba : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/090962b6fd43469681e6b10b8dbc5c02 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043985388deb45788a5a62b51614f2c3 : SUCCESS in 3m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d0cf9918bcd344eaaaab638cc81a4999 : FAILURE in 3m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df96f3d359c446d6b1a5a1a8c087c2df : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5e94dac120952c0ec34bb95140df4190ec4fa710","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-11 11:35:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"96244181d33b9db988eaf7b41314ff37cb3e4450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:41:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e93a793a529484e9b77985576a20e47 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c63c4b25d8634cfcb29651277da18919 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68b6eb7c15e45008faabf202044953a : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/144cc02954984454bd1941acd7ec2fb1 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09401c7850c94fb5abfe8e056fea3517 : SUCCESS in 3m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8c025b610f0f4813a318e927f38b299c : FAILURE in 2m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4c6d798f99042d690fc73183c2165a7 : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9f8cd5ac39d7b3c1736b9db563315627f3a9df03","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-02 11:55:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73facc462761cd0f1e72b4d38208d6aa9f93c14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 12:05:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/332124d6f4234e8f94d3bf332951223b : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee3b9bf520684cf5aeec3f4d21e0a5fe : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/300d20a03290497088411c7fae30e076 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be104acdf4784a57b6185492b9fcb4e7 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7bec1273434679ba962509b5ac2bf4 : SUCCESS in 4m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0fb368ac7c9b49f29fda77819d418429 : FAILURE in 3m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8dd4cdddc7484d4aa243af89b89cc031 : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e5df94e52807b2e7cb88f6610a292d88d724689a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-09-08 10:29:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"088395081d273dba649d7cfa7043e3f3f9322110","revisions":{"088395081d273dba649d7cfa7043e3f3f9322110":{"kind":"REWORK","_number":1,"created":"2021-08-11 11:21:03.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/64/804164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/64/804164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/804164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/804164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/64/804164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/64/804164/1"}}},"commit":{"parents":[{"commit":"76e26bdb2f6cce7a2c942dd43f56d2ae1c621296","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/76e26bdb2f6cce7a2c942dd43f56d2ae1c621296"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-08-04 06:28:15.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-08-11 11:21:03.000000000","tz":0},"subject":"Update command help information","message":"Update command help information\n\nupdate openstack allocation candidate list help.\n\nChange-Id: Iad530edaec39545a5573196f4c33ca1b29808fd7\n(cherry picked from commit 7aed6dcd06c17a80b621d53fc9a690f3028c252e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/088395081d273dba649d7cfa7043e3f3f9322110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/088395081d273dba649d7cfa7043e3f3f9322110"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
