)]}'
{"id":"openstack%2Fosc-placement~807556","triplet_id":"openstack%2Fosc-placement~master~Ic0e6e981d6602a76935f6bc6d9ffb0a707a5b1a9","project":"openstack/osc-placement","branch":"master","topic":"bug/1942740","hashtags":[],"change_id":"Ic0e6e981d6602a76935f6bc6d9ffb0a707a5b1a9","subject":"Fix allocation show / unset with empty allocation","status":"MERGED","created":"2021-09-06 09:51:45.000000000","updated":"2021-09-06 14:22:56.000000000","submitted":"2021-09-06 14:20:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807556-bug/1942740","meta_rev_id":"3994f0c84dd08929678a588445566b09e2de3cc7","_number":807556,"virtual_id_number":807556,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 14:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":2,"date":"2021-09-06 13:42:19.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-09-06 14:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-09-06 14:06:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-06 14:22:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 09:52:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2021-09-06 09:52:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-09-06 09:52:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-09-06 09:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acecdd616afb321c31edd330a40932487b127c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-06 09:51:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"255cf207556c9f96dbba5a0911e22a81bf94b7ca","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-06 09:52:04.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ad5bdd862208a037f4967566081f66c0fa1157e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:59:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbca6b47a3514f19a41647b69b2f81b2 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f104e3a409b94addaec707fd33f441c3 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35d48875e6734fcf98ce1a1cdd1d230f : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/722842ffd97f4dfd9a634a01c78f801f : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc0615bfc234bc197a29fe65b1ad8d1 : SUCCESS in 6m 24s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d8677b83351c44e299f030de0d2f8c29 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e0d99774aaa4ccead024684002b3991 : SUCCESS in 4m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3a50648329954e27bda0f98b8f641e3f : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ffc9c894e8b34bad995920f522e36cfc : SUCCESS in 4m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15fe516642f9463b81eee79cb8f9ece1 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68d59f0b0e96ab5b163c76ffbbfc3236338cdb3a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-06 11:56:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b2b183a12f327652e2a1eee2c5146088cdbd308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-06 13:12:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9dd9e21add9218164a34dd658736762e722f3235","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-06 13:12:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26f49f0e31128b7d6fdfc26e2db871e413736ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 13:24:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984a1ff17ade49e29c1bbdf1d972f712 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ccaab34087c425bbf00972c863b1b58 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9add5157b04b4555aa2c4448f1da83e1 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/693d239da5e145219208efcce0dc667e : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56eb6903684488880351f81076b09b6 : SUCCESS in 6m 43s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/66b16eadac4943f293ab648778e3816a : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/083398901ed343cebd6d0c4353b41b54 : SUCCESS in 5m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/05bd5fa2dd0348d7b2e724cc0d4b9ce0 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae32d692e9e448d181cc61d7accbbde1 : SUCCESS in 4m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a7d40174ce04510bb370f00fca48586 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8154d3a48b7fe396c3590d2d819849dbdf8c81ca","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-09-06 13:42:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLGTM, quick nit in the commit but the actual fix is correct and well tested.","accounts_in_message":[],"_revision_number":2},{"id":"6feb6b4d19038f166d26f593fb2c83ca79aa6d68","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-09-06 14:06:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c745caecfe44876fbd1eb5df461430642f9360fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:07:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e75036ae80e6f49a700473602b5a21d2cd7511c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:20:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3b52d1aa094a5e9b8ef77f207bf2d8 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5010a9c9b5fc48eb83e160cc1991d70b : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2541bce46ddc43ed8c6409984ed68614 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8654c96cbe2d4e508ab9eb4fd579391c : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e0c793d49e64ab791f7ee1fb1618fc4 : SUCCESS in 4m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/57236246033542b8a03221d0b5e971cc : SUCCESS in 6m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35459a54262142fd970ed98f4677c9c3 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4bc25456c9e68fc57cb863e64a96c70d955be3ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:20:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5aff0818dcf18e39760e4edb871fca6d120314ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:22:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e49d2cc1074a71b264afffb089b2bd : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ea8f1b05d4aa4a6f144adff08771f5b84bda479","revisions":{"8dee42f8b906417ed16a32c904cd02cbbb191703":{"kind":"REWORK","_number":1,"created":"2021-09-06 09:51:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/56/807556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/56/807556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/1"}}},"commit":{"parents":[{"commit":"b797488dd762b09df609eec4c5598f036fb695e5","subject":"Repro allocation show bug with empty allocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b797488dd762b09df609eec4c5598f036fb695e5"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-06 09:49:51.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-06 09:49:51.000000000","tz":120},"subject":"Fix allocation show / unset on empty allocation","message":"Fix allocation show / unset on empty allocation\n\nAdd a guard for empty allocations in the allocation parsing code of\nallocation show and allocation unset.\n\nCloses-Bug: #1942740\nChange-Id: Ic0e6e981d6602a76935f6bc6d9ffb0a707a5b1a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8dee42f8b906417ed16a32c904cd02cbbb191703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8dee42f8b906417ed16a32c904cd02cbbb191703"}]},"branch":"refs/heads/master"},"0ea8f1b05d4aa4a6f144adff08771f5b84bda479":{"kind":"REWORK","_number":2,"created":"2021-09-06 13:12:10.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/56/807556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/56/807556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/56/807556/2"}}},"commit":{"parents":[{"commit":"087dd64dd2b92ea0d3a1d96cec543e2eb6ca813f","subject":"Repro allocation show bug with empty allocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/087dd64dd2b92ea0d3a1d96cec543e2eb6ca813f"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-06 09:49:51.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-06 13:11:47.000000000","tz":120},"subject":"Fix allocation show / unset with empty allocation","message":"Fix allocation show / unset with empty allocation\n\nAdd a guards for empty allocations in the allocation parsing code of\nallocation show and allocation unset.\n\nCloses-Bug: #1942740\nChange-Id: Ic0e6e981d6602a76935f6bc6d9ffb0a707a5b1a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0ea8f1b05d4aa4a6f144adff08771f5b84bda479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0ea8f1b05d4aa4a6f144adff08771f5b84bda479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
