)]}'
{"id":"openstack%2Fosc-placement~828545","triplet_id":"openstack%2Fosc-placement~master~I38ff55bdd072f3a9c1ed03e28192d045cb4096cf","project":"openstack/osc-placement","branch":"master","topic":"add-zed-python-jobtemplates","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-06-30 14:37:56.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-06-30 14:37:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I38ff55bdd072f3a9c1ed03e28192d045cb4096cf","subject":"Support microversion 1.39","status":"MERGED","created":"2022-02-09 14:54:51.000000000","updated":"2022-06-30 14:39:35.000000000","submitted":"2022-06-30 14:37:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"828545-add-zed-python-jobtemplates","meta_rev_id":"d48728d7e34dcd94684b115cb9f4bacca324811c","_number":828545,"virtual_id_number":828545,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 14:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-06-30 14:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-01 14:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2022-06-02 15:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-06-30 14:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-06-02 15:32:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 15:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-25 15:07:24.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-06-01 14:14:06.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-06-30 14:27:16.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"6987ac1b6923efc4806b961dba30e44e8baa94bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-09 14:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86eaef0836c60c73182ac2950d970a085767279a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-09 14:57:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a2d8831312d806aca6906b4cc8915eeb9d7c31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:43:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8450136bc14443aab2e9fad40aa16fb : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56ada65a740040e4acfb04c66cacef8c : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38877ce54ef411c92281f92279c8d46 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3049ed5a0ef042c7a93ef93a5a8ac285 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53c0a6894f24547822a5b3d0251723b : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e062f58de49437aaaff9d0cb19499ee : SUCCESS in 4m 29s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/8a5db41147504f15ae01ef7f4322be4f : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8434e7f02f04357961c2ffb360ce7f2 : FAILURE in 5m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2a81b3e50e6846589d87f0c152fbbceb : FAILURE in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e5708ffd728470b8962154058a854e0 : FAILURE in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8764192bf5804131aa5efeb808036670 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d3036382c7a4d5eed87f9da1b2fe946fb617d056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-09 16:54:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b8dc1a0f36758170b08967c5746f249f9371f06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 19:15:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359b62c5f77340838ca67bf3a79341bc : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e6a28ad3b78404f843d74458b5d5028 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e51baad7d284f41ab920cf7e0ebfd3e : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/291fbe7cdc914edebb36f6b047f27343 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b0f73b52c54b819e1a805676236a66 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5a2a38d78145019819c466ddef1fd4 : SUCCESS in 5m 57s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c12dc34e27004783a825f558a9926f43 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63093562cad549b79c38c403cf554a7b : SUCCESS in 6m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a3d241d55a154454a0813e2b8661b8c6 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8b07e0c71974506adb2a218f6220d8d : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81a10e10c1e84961a3946a9603e94710 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"af848401d6b5984b87818324ecb0331a0a720362","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-02-25 15:07:24.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"7b86760c573f3b0765dd59f2e0f207ac85e8c586","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-03-01 10:50:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c32c33118f3eaa921b642f4dd79a840b648a8a62","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-03-01 16:34:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2e9c2c572cc45c37eb296b08ecc3262f7b7c331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-31 12:47:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21bbfc7ce22820714621dc555437f7bee41ee9d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-31 12:48:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b2a9b433c7bf4823c834606cc08d0d409d0ebad7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-31 12:48:28.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e264095b5b680225c1da59e6b9fe00dc0d9e882a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 13:30:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1592decb23b4a95bd0b24f433f1d5f5 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d1009ea65d43f8b5ee93db2e87e4d0 : FAILURE in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3441e5b930e47edab82c05d38c6227a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbb76416e6846d79160b998e38f47af : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14e1c2ac4de84181a20a8cf40d1fb806 : SUCCESS in 3m 12s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/356d88306af34d2ea37269aa3ae0cba7 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f927c96df6f41c4816a562a91dc0f6e : FAILURE in 3m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7187e49485f84f6ead4395a66b5abe21 : FAILURE in 2m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e21a0aa64634394a6400717038952a3 : SUCCESS in 4m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3312ffd6c4384e54b9b1dc20df661d8f : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9831968fc3c2893600575c7c7eedcb1ea2c06723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-01 09:32:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"70067a02dc6cd162d1a7b513fca11c54dead3919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 10:01:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ace5bb0c494261a5ad52c96407472c : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c42609791895449e8022e6875be20569 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a6cd0c557724975956a1de889d437b3 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a86f4282c6cb45748ec78596f8deb35d : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a38acc80db41f999025df61a573d99 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22ddaf6aefe143d48e6dd57435208b01 : SUCCESS in 3m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9e3cf2d8bfee4129816442f2523676d6 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12bf17c27ac7438dbb612b56cc26a374 : FAILURE in 3m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2884b47c698149f889621055f5d6854c : FAILURE in 3m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28bbdbb91afd4f059307148ba6959ae8 : SUCCESS in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b771eee39274e31bb8039b5610e3b92 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":5},{"id":"8a8240ac789c4c32b78879550bc96e064cf5ee13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-01 11:29:09.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b0f7e4473759172c21073d7707bdb055eddbe64d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 11:47:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4c185444354f288b6c5bd5c4c51fc5 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bc80bfce8fb43f186ea597c339da1dd : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/872bac1dd1f34e59b5f0febc138675fa : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f5dd4893acb4289a5972fd540f72bd0 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a54888b54454d57b28ee7df8e69de05 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c85b39b9819415286158314da52771d : SUCCESS in 3m 30s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/2c73b5a1a2e8456ebffc0c70ee86b09c : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6391d9d9c4114ed0aa9530e9907bccbd : SUCCESS in 3m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4bf394c0b96434991fa749fc6ecebf4 : SUCCESS in 3m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37a978f27d8246518fe5d86e06666796 : SUCCESS in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f7f3b6cfa1499a27bc226fd3a274ec6f1f1b6688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-01 12:09:04.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fbd0430db659460805f1ab292bf4e8f7784290ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 12:26:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5ca72a1ab54f8888438de75e7ef429 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/665e6735ff354787abf5521a207f39b8 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fbc4e4cf42445fcab46bee1b1c095c5 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72fc94ea5cb24693859a09ad854dbd5b : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f847800c2fc741eea64febc3eba66205 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ef3d78da934cfaa746395f73ffbf2d : SUCCESS in 3m 00s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3ea9c6f0219d415e86f1aa52044bcd06 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f02d1dd496bb4e5d8095bbc8001ef0fc : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56dd05d50d004e239259c641f4418e63 : SUCCESS in 3m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e95db6e3b4af4390a52782932588a303 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"272d3fbcc8420f65db838cad248b4533931e0daf","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-06-01 14:14:06.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"33599c37ad7551f23f87a5cabb49e9816af4fd9c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-06-02 15:32:30.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"904eb0904233cbbee7081e443db161da394bae06","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-06-30 14:27:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6ad7b27f55bb991b4e7ab2cf531af06f855920be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 14:27:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0710cc42f74e2aed0da252f592b3f990f82a1de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 14:37:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394ead9c08d44c099defe12a5fc67986 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111321f66777498ca4386c81d17e90fd : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65f2192267b443e29e97129744b5e6de : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439d42c3c9d54371991f0ec603be7883 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a976ecd6d6c4742999e39665eca1044 : SUCCESS in 2m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0dc1f62061a548bea9ada8839fc8a521 : SUCCESS in 4m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63c953b991be4c84aa80d6f8e1e3f72b : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4b4e713e48f425c91670b101221cb7a : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f643ed0e812191b149cc84d1396828d52875f5af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 14:37:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d48728d7e34dcd94684b115cb9f4bacca324811c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 14:39:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1db7f50bde4ba6b9d3dabe16d6a016 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bd2e774d2a443f7ab2d1790cedf8e7a : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9545623054fda8313cde1253d9af3fffcdaa3949","revisions":{"a8b45352da6b088534e51b8782d79b2869c96aab":{"kind":"REWORK","_number":1,"created":"2022-02-09 14:54:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/1"}}},"commit":{"parents":[{"commit":"0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0","subject":"Updating python testing as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 14:54:31.000000000","tz":60},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a8b45352da6b088534e51b8782d79b2869c96aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a8b45352da6b088534e51b8782d79b2869c96aab"}]},"branch":"refs/heads/master"},"4e0cb071b20162fcbe3ceffce81b8e02d7309367":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-09 16:54:15.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/2"}}},"commit":{"parents":[{"commit":"0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0","subject":"Updating python testing as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 16:53:30.000000000","tz":60},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4e0cb071b20162fcbe3ceffce81b8e02d7309367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4e0cb071b20162fcbe3ceffce81b8e02d7309367"}]},"branch":"refs/heads/master"},"ec1d9a28bb2f13bbb5ea7758c3c29f8e2630f89e":{"kind":"REWORK","_number":3,"created":"2022-05-31 12:47:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/3"}}},"commit":{"parents":[{"commit":"0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0","subject":"Updating python testing as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c24adb61ebbd7e4a3bf7a06f13ae0bdb46746d0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-05-31 12:48:08.000000000","tz":120},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ec1d9a28bb2f13bbb5ea7758c3c29f8e2630f89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ec1d9a28bb2f13bbb5ea7758c3c29f8e2630f89e"}]},"branch":"refs/heads/master"},"fe566ec2d731d25bd391be4b7188bbb6ec6d4168":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-31 12:48:05.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/4"}}},"commit":{"parents":[{"commit":"6c1ebc9d42fb51cf51f492c8b05b09e844413d2b","subject":"Merge \"Remove usage of six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6c1ebc9d42fb51cf51f492c8b05b09e844413d2b"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-05-31 12:48:44.000000000","tz":120},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fe566ec2d731d25bd391be4b7188bbb6ec6d4168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fe566ec2d731d25bd391be4b7188bbb6ec6d4168"}]},"branch":"refs/heads/master"},"57e65eb26f885247fbd9965a882304fd33b78a3f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-06-01 09:32:28.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/5"}}},"commit":{"parents":[{"commit":"b49712bdc291f448450c4aa0ba8e6e75ed9e35b4","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b49712bdc291f448450c4aa0ba8e6e75ed9e35b4"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-01 09:32:28.000000000","tz":0},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/57e65eb26f885247fbd9965a882304fd33b78a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/57e65eb26f885247fbd9965a882304fd33b78a3f"}]},"branch":"refs/heads/master","description":"Rebase"},"34adca82d565ec2c220a7857190adf0373ed83fd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-06-01 11:29:09.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/6"}}},"commit":{"parents":[{"commit":"fc3470630799045e1d3ed2b4bfee9f7cf33d4505","subject":"Drop py36 and py37 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fc3470630799045e1d3ed2b4bfee9f7cf33d4505"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-01 11:29:09.000000000","tz":0},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/34adca82d565ec2c220a7857190adf0373ed83fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/34adca82d565ec2c220a7857190adf0373ed83fd"}]},"branch":"refs/heads/master","description":"Rebase"},"9545623054fda8313cde1253d9af3fffcdaa3949":{"kind":"NO_CHANGE","_number":7,"created":"2022-06-01 12:09:04.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/828545/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/45/828545/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/45/828545/7"}}},"commit":{"parents":[{"commit":"91017972278a87684d8ad10d2ae0f8f57aa19251","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/91017972278a87684d8ad10d2ae0f8f57aa19251"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-09 12:39:01.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-06-01 12:09:29.000000000","tz":120},"subject":"Support microversion 1.39","message":"Support microversion 1.39\n\nThe CLI now support the following syntax both for openstack resource\nproviders list and allocation candidate list commands:\n\n  --required T1,T2 --required T3\n\nand it means (T1 or T2) and T3.\n\nIn the allocation candidate list command the above can be used both\noutside and in a --group context.\n\nStory: 2005345\nStory: 2005346\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/826719\n\nChange-Id: I38ff55bdd072f3a9c1ed03e28192d045cb4096cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9545623054fda8313cde1253d9af3fffcdaa3949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9545623054fda8313cde1253d9af3fffcdaa3949"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[]}
