)]}'
{"id":"openstack%2Fosc-placement~842386","triplet_id":"openstack%2Fosc-placement~master~I1d8de7f77c472e3a446b363d2e0efe6977514e30","project":"openstack/osc-placement","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-05-18 18:00:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1d8de7f77c472e3a446b363d2e0efe6977514e30","subject":"Remove six","status":"ABANDONED","created":"2022-05-18 16:09:41.000000000","updated":"2022-05-18 18:00:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4e1c6a69b8d3e96e4ca907dec48dae33338de83","_number":842386,"virtual_id_number":842386,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-18 16:23:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 16:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 16:47:13.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"}],"messages":[{"id":"a27d482b9222b866bd04b172b44dee9f08845622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-18 16:09:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b0ca08a40837b9c4edcc259b07e6b3d32454709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:23:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2d1248db3e4a248250bdde39662938 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/916ffde5010448e3b88fa69d5009a495 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82167c53dcf447138c2254f17636a360 : SUCCESS in 4m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/affcb01ca39e493f9feb926ea69a50f0 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb681a3fb7843e39bad1cbb92a120bb : SUCCESS in 6m 07s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/40d203d0cf9a4d9497253273f841cc49 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fbd773492f4452eb49f50660edcff8c : SUCCESS in 3m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/28c2f7ff752f493abaf5f8fb3e2fcdae : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5177653b261543eebcf70257341c587e : SUCCESS in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3bcd0c614914f70ad6509a4d098630d : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"163d24cda3618b81b7108b53458976e66eca7cdd","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-18 16:47:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4e1c6a69b8d3e96e4ca907dec48dae33338de83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-18 18:00:47.000000000","message":"Abandoned\n\nWhoops /o\\ Let\u0027s go with the existing changes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aca33346acf33f40cbf202c5274a98cd354143b6","revisions":{"aca33346acf33f40cbf202c5274a98cd354143b6":{"kind":"REWORK","_number":1,"created":"2022-05-18 16:09:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/842386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/86/842386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/86/842386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/86/842386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/86/842386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/86/842386/1"}}},"commit":{"parents":[{"commit":"7f77e98abd4387f975ba7bf9526043a8b89068d6","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7f77e98abd4387f975ba7bf9526043a8b89068d6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-18 16:09:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-18 16:09:25.000000000","tz":60},"subject":"Remove six","message":"Remove six\n\nNo need for this in a Python 3-only world.\n\nChange-Id: I1d8de7f77c472e3a446b363d2e0efe6977514e30\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aca33346acf33f40cbf202c5274a98cd354143b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aca33346acf33f40cbf202c5274a98cd354143b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
