)]}'
{"id":"openstack%2Fosc-placement~868359","triplet_id":"openstack%2Fosc-placement~master~I1acceed2709ea855b152fe2bf8dae8c9eaf339ca","project":"openstack/osc-placement","branch":"master","topic":"force-tox-v4","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-05 11:03:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1acceed2709ea855b152fe2bf8dae8c9eaf339ca","subject":"DNM force latest tox","status":"ABANDONED","created":"2022-12-21 16:02:50.000000000","updated":"2026-08-05 11:03:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"02bb5006d5ae46e8722b32404eccfbef1e4246c3","_number":868359,"virtual_id_number":868359,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-21 16:16:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 16:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cde6d48950d785ce6af2f2a2edcae19a2a468ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-12-21 16:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b2f757775e978f25390db95846ed90280b486ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 16:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e09e8b4d5c648568f0cb9a5222d6170\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff8f37b523d145bab60666fd5c06a0c1 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f4dea33040471caf4467e3be07ae02 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/993ceed3e8b44b7e946556da448bdb13 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6f751950e124e648edaa3fc57bda806 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/910e6f31d70147db8bacab46e7e95055 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06981761aaef421d8225700d52441332 : FAILURE in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa1593d532e64d2ba6a7a87ea29a64b4 : SUCCESS in 3m 00s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0477a684b3f64c80b29540e7e771a9fe : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5a2ebaac2584a98a6ba0bf4571c2f2a : FAILURE in 3m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3bd8ff0258440edac231f4076695dcf : FAILURE in 3m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc0819cf45ac4f879e42fb14056122df : FAILURE in 3m 32s\n\nWarning:\n  Comments left for invalid file doc/source/cli/index.rst","accounts_in_message":[],"_revision_number":1},{"id":"02bb5006d5ae46e8722b32404eccfbef1e4246c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 11:03:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6706ad63d3ea8e2c00d9af12bf2c3a84aa668975","revisions":{"6706ad63d3ea8e2c00d9af12bf2c3a84aa668975":{"kind":"REWORK","_number":1,"created":"2022-12-21 16:02:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/868359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/59/868359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/59/868359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/59/868359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/59/868359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/59/868359/1"}}},"commit":{"parents":[{"commit":"e3268569658b262792f481767b4128f2cb4b10ef","subject":"Update gate jobs as per the 2023.1 cycle testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e3268569658b262792f481767b4128f2cb4b10ef"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-21 16:01:51.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-21 16:02:46.000000000","tz":60},"subject":"DNM force latest tox","message":"DNM force latest tox\n\nThis change will test osc-placement against tox v4 to check compatibility.\n\nChange-Id: I1acceed2709ea855b152fe2bf8dae8c9eaf339ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6706ad63d3ea8e2c00d9af12bf2c3a84aa668975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6706ad63d3ea8e2c00d9af12bf2c3a84aa668975"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
