)]}'
{"id":"openstack%2Fosc-placement~869451","triplet_id":"openstack%2Fosc-placement~stable%2Fyoga~Iaf749a18f060551833d0673f5f965aea7c607d8d","project":"openstack/osc-placement","branch":"stable/yoga","topic":"tox4-pin-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-10 19:08:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaf749a18f060551833d0673f5f965aea7c607d8d","subject":"DNM: test tox\u003c4 pinning in stable branches","status":"ABANDONED","created":"2023-01-07 02:24:14.000000000","updated":"2023-01-10 19:08:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":869517,"cherry_pick_of_patch_set":1,"meta_rev_id":"42206e0a96c4066522dfd29b7b8e309a4b2c6cb3","_number":869451,"virtual_id_number":869451,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-07 02:39:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-07 02:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c026b6141c855e0f43bc77455b94fba984074619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:24:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"a504f0aa3567303fb191545ba8ac0c89ea549179","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:24:25.000000000","message":"Topic tox4-pin-testing-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"087f2b64441c2120a04037b0ac6bf99427fad9fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:24:31.000000000","message":"Topic set to tox4-pin-testing","accounts_in_message":[],"_revision_number":1},{"id":"58c32c45577771015043607cd72d628ce6b5f2d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:39:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2836154bc95147eead84ed35969d6588\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4a69a527444cfc8d535afbb0ac9360 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d976595b9ad4e7ea28123753ac005d9 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44add4d92fa345f4b04c57f3a9d8ebb5 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0294c1779696419fb4e8fb8d65dfefdb : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7486b8566c247d0bd018ee5eb8321c3 : FAILURE in 3m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ee6bea99326b41a1927947afd9a0d285 : FAILURE in 2m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75 : FAILURE in 2m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/826966a02b4543f0858d5c160c925c28 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"42206e0a96c4066522dfd29b7b8e309a4b2c6cb3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 19:08:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aead438c75a94edfa5dfd3d05de2749d2ebb187d","revisions":{"aead438c75a94edfa5dfd3d05de2749d2ebb187d":{"kind":"REWORK","_number":1,"created":"2023-01-07 02:24:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/869451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/51/869451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/51/869451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/51/869451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/51/869451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/51/869451/1"}}},"commit":{"parents":[{"commit":"07472fa54b8e215ca4a5819e21e036d01102a70a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/07472fa54b8e215ca4a5819e21e036d01102a70a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-06 23:59:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:24:14.000000000","tz":0},"subject":"DNM: test tox\u003c4 pinning in stable branches","message":"DNM: test tox\u003c4 pinning in stable branches\n\nChange-Id: Iaf749a18f060551833d0673f5f965aea7c607d8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aead438c75a94edfa5dfd3d05de2749d2ebb187d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aead438c75a94edfa5dfd3d05de2749d2ebb187d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
