)]}'
{"id":"openstack%2Fosc-placement~869754","triplet_id":"openstack%2Fosc-placement~stable%2Fyoga~I4e3e5732411639054baaa9211a29e2e2c8210ac0","project":"openstack/osc-placement","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-09-29 11:51:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e3e5732411639054baaa9211a29e2e2c8210ac0","subject":"Use pypi released version of placement in functional tests","status":"MERGED","created":"2023-01-10 19:08:39.000000000","updated":"2023-09-29 11:52:34.000000000","submitted":"2023-09-29 11:51:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869754","meta_rev_id":"dd4d0ad4fdb330891f04bccf5508731b6e248287","_number":869754,"virtual_id_number":869754,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 11:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-09-29 11:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-09-29 11:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-09-29 11:42:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 19:17:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 18:51:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-09-29 11:10:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"04c9dfbe54dab840fe9ae2c7389960c7a940bcf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 19:08:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f581ea2eeb3507b918d9d472c6adfca4db34e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:17:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11dd823575f448789154cc9f0d2f818f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ee522592bd4764bec1525c2bcc40a7 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffdae1cb0bff4b2eb6b21a20a710eac6 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5ddb557d57341c99c8c914210887fab : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824f6879126047399727487400a07178 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5417c6b631e94527bf2eb6d90c5e7099 : SUCCESS in 2m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ca9a3148cfe4577b7d46936f44c5963 : FAILURE in 2m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e284035658304629a5789a2d0d40d42b : FAILURE in 2m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92791a247a784563b2dc609e7b0ee677 : FAILURE in 2m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45099e3d56cf45f0941cc3135853ba67 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4ba2beeb2127ca42418d5d881db5324e54bc4982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 19:28:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cb6cd6f2b6a735eea4313c26979fedf4aea78da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:34:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6abf13c1fe8f4978bc30ed6f0c03cd09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18822fcf53148c9920330a97cb08f9f : FAILURE in 2m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e52c0e9a9a4b85b14527ad1836bfda : FAILURE in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8accfbc5a7842168e189d7a8f000b8b : FAILURE in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a71b76f58294dc3a3f22ec0d2db9bbe : FAILURE in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97b7103bae442f2b286d4ad4081c2a4 : SUCCESS in 3m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1de6e6ffd8145019f0a2936b819683f : FAILURE in 2m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/23e08d12ea9c4e5798788cbf14e794fa : FAILURE in 3m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3cd22f6f8c4b4a048a92daa506ce31eb : FAILURE in 2m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30b3a9cffac1416f94f8496652f03865 : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"09c4e610bf09cdbd8a9adb31eac1c65968efc75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 20:24:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0746a39b81cfa0ae36efca6aa00598bb47996431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 20:37:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35db83d7ab4140498b32a049d13885b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa22cce5ef5a4bb5af379dffee3e719d : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c2e508baeee48618ab084241b03a61f : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a44c0da4d772448a93a6fab77fdd8c4e : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae744648079b462fb8b3b53f9a336db4 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d68f54325d16443f9351ca5b3e718b61 : SUCCESS in 2m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ab6154410dc420cab536e054b73bae6 : SUCCESS in 4m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/80ebbe5d1e6f4bb4a78001c4473f01cf : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e394ce1b09e4d56b13762e8dff0b7ad : SUCCESS in 3m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad98f1e8f0f44f3087b8651a5e0c6585 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"99323153c8d003f9ca2c80137baf686b32f6fe28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 01:10:34.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b366ffaa513bd26eb21521f8abe004e5a0e0a1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 01:25:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3aca7da9523479fa3eb1f0887c0252a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b526291a9e574d05bcf55cac5e8c71bc : SUCCESS in 2m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5e170db701d4174b0d6f2a43ebd63e4 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c8550225074c9881390d1871b13c96 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b2c58dbc72473c8510637ed73c3822 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e35eb5ea7a43aca389f474001f6810 : SUCCESS in 2m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59fbbc46e62a49b8bf2bdf0c9fe87e9e : SUCCESS in 3m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bfe291bc58de4e97a772687cdf410cee : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e4f07651cd349339d13c3077d1f0724 : SUCCESS in 3m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78581bf4613548ed9e8ed55a3d5ec63a : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c837b49f2843a7877afb516ba83bee5a175b5e6d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-13 18:51:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b3cce7ed4817a8b393d90085050e511761a8aa6c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 11:10:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f05a01d363614c7649940fc7499b14d4fa58a889","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-29 11:42:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"26c93e46b5a048935469e6785a0a7b6256014470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:42:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f587e743bc6a664d5429a1e1d41cc2b50cee394a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:51:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1eb5f5b3a544bee9672151ae0aaba5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f5933dfe5e41218aca33b76dd44922 : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32aefd85ac2a40b0a6d1505722025345 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b39af49c9bd47debfffe83a23704461 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626f824aa26a4e4c9f03e6d864399cef : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d00a339524a49cfb3e652604c929b9f : SUCCESS in 3m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/043a2a82cfa04a3183c755d1d21b4a5f : SUCCESS in 3m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/73e55a174a7f4b6cb917a22c46ff735e : SUCCESS in 3m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24b2a4ee6b3547d8aaf9da37e5b07010 : SUCCESS in 3m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0533c64477b8457bab9bdf7fe1bb7e0e : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"002980f050d5372b809324c467ef5dca98b72b8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:51:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dd4d0ad4fdb330891f04bccf5508731b6e248287","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:52:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47346c54a1b84107a392ee20c9edf35a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbff2c6652854a1ab77641ebbf96bd8b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ed75c52510741e98469495e119a60e5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d2c170d778c7aadccae602b9e7e95144b9d888ad","revisions":{"55e468c0b457b812b61ccb6bb18eb4e7f9c52d52":{"kind":"REWORK","_number":1,"created":"2023-01-10 19:08:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/869754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/54/869754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/1"}}},"commit":{"parents":[{"commit":"07472fa54b8e215ca4a5819e21e036d01102a70a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/07472fa54b8e215ca4a5819e21e036d01102a70a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 18:52:15.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:08:33.000000000","tz":-360},"subject":"[stable-only] Use stable branch version of placement in functional tests","message":"[stable-only] Use stable branch version of placement in functional tests\n\nosc-placement functional tests are using the master version of placement\nwhich is failing due to the constraints mismatch\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is original patch which started using the placement from master\nsource which make sense on the master gate but for the stable branch testing\nwe should use the stable release version of placement\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit 11d57ab51ded21b106342c40848a60a1dd74c31c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/55e468c0b457b812b61ccb6bb18eb4e7f9c52d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/55e468c0b457b812b61ccb6bb18eb4e7f9c52d52"}]},"branch":"refs/heads/stable/yoga"},"1d878cdd930f826a8c68f3ae44dac7bcf6ef3256":{"kind":"REWORK","_number":2,"created":"2023-01-10 19:28:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/869754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/54/869754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/2"}}},"commit":{"parents":[{"commit":"07472fa54b8e215ca4a5819e21e036d01102a70a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/07472fa54b8e215ca4a5819e21e036d01102a70a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:28:15.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit c93e91665981eaacf38b677cdb33a4380c18bdcc)\n(cherry picked from commit 0f1a2cc3dc95685074e7e0a0d5e17afadf5dc1f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1d878cdd930f826a8c68f3ae44dac7bcf6ef3256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1d878cdd930f826a8c68f3ae44dac7bcf6ef3256"}]},"branch":"refs/heads/stable/yoga"},"4dfbfa14c687dd4c4bbf11fa65939b84b6d68baa":{"kind":"REWORK","_number":3,"created":"2023-01-10 20:24:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/869754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/54/869754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/3"}}},"commit":{"parents":[{"commit":"07472fa54b8e215ca4a5819e21e036d01102a70a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/07472fa54b8e215ca4a5819e21e036d01102a70a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 20:24:11.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit 90ecfa92f030a6370658430ca3bbc6273f7ea487)\n(cherry picked from commit 3f2a80b68331d774947d0716151daaf0bfb94ac1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4dfbfa14c687dd4c4bbf11fa65939b84b6d68baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4dfbfa14c687dd4c4bbf11fa65939b84b6d68baa"}]},"branch":"refs/heads/stable/yoga"},"d2c170d778c7aadccae602b9e7e95144b9d888ad":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-03-11 01:10:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/869754/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/54/869754/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/54/869754/4"}}},"commit":{"parents":[{"commit":"07472fa54b8e215ca4a5819e21e036d01102a70a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/07472fa54b8e215ca4a5819e21e036d01102a70a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-03-11 01:10:26.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nDuring this backport, removing basepython changes are not applicable on stable as we have\npinned stable branches for tox\u003c4.\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit 90ecfa92f030a6370658430ca3bbc6273f7ea487)\n(cherry picked from commit d4036554f044f548d6ba460db0114c0fd7469f7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d2c170d778c7aadccae602b9e7e95144b9d888ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d2c170d778c7aadccae602b9e7e95144b9d888ad"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
