)]}'
{"id":"openstack%2Fosc-placement~869755","triplet_id":"openstack%2Fosc-placement~master~I4e3e5732411639054baaa9211a29e2e2c8210ac0","project":"openstack/osc-placement","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-17 17:50:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e3e5732411639054baaa9211a29e2e2c8210ac0","subject":"Use pypi released version of placement in functional tests","status":"MERGED","created":"2023-01-10 19:23:49.000000000","updated":"2023-01-17 17:51:30.000000000","submitted":"2023-01-17 17:50:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869755","meta_rev_id":"0ad13d51df93dae3b81c1649001d5d082c7cc247","_number":869755,"virtual_id_number":869755,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-17 17:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-01-13 21:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-17 17:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-17 17:40:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 19:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 21:49:25.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-01-17 17:40:19.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"5bc54edd11f5d96731520609dcaba724d65bc0f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 19:23:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1451f2fe0d8fd872ea66834acb257519fbbf121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:32:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6670dc5470a040ffae7fd7e3bc1394fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d14f3bada534da9bc4ba8eaca3757cf : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f68dc92eca4db39086c7dc988b7c39 : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29b8350307294b2186d2a66d9d5ac562 : FAILURE in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ee99d447ab4807be2e79767f98451f : FAILURE in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eb5d72a953d4833b492ac59e27f1a8a : SUCCESS in 3m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/650f33dc492c42d5bc1e6d0c98185230 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a0c8ed6dbfe48c7b85c6491282380e4 : FAILURE in 2m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3135bf257ef499681f18fb4a6ab9f3c : FAILURE in 3m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f41dd1058ee4a56887f9d237f8969d0 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"942c374b242354b26c45b2598c16f1abcf89f690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 19:58:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"82114df66eb9798f43ee04d04c8e8007c6f87c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 20:06:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4780295e3b84beaa7ae01369e6c275f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21792cda5a084f0895e2a338c6369e90 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80201a7d1b1a4773a9a1902093555fbc : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bb35ba5b2cc4eaf97bb944e6f310563 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8f80c18153447593af0febeeb3af3b : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66a26548dc964e79b9b3ee666aaf212d : SUCCESS in 3m 11s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9d307607d87a419aaa1252e59675d151 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b922d26915a64e90b29ed5f8ed13cfa7 : FAILURE in 2m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d77cfe3504f44111b0fa45d96cc95d0a : FAILURE in 2m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/138f7fba14154114a438066b26b19ba2 : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d90d53c08acf199520b1b1956972ef098a62db9d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 20:46:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1e4b9977f90de185f49c2fcf678ac39290f8bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 20:56:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc7969a998e03713b23fc2d35ea1a4fc8e175d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:07:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/335928beb3fa4c63b4e62f0b262a5cef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/520f4b07f6de4bc3ae4ada83a02f0be4 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/007e8083ff254cadb5482e60cb1a89b8 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99361abff4f5408fb1b99a56dfbccf38 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4120fc2287a5496a9f27f748f05eb5c3 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebccaf46e9964aa68d495c45b6f5bb30 : SUCCESS in 2m 32s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/da666a57d64845ecb2ad092a69ebf996 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3879af192d814e44985a518d08e7dbbc : SUCCESS in 3m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51b9be6a3628431d8558689235f00743 : SUCCESS in 4m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4f3c4255539465681058c684bba2af4 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"577ea395d1634e33470d18984790228de30ac003","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-13 21:49:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f665fd8121ecc269e7b16b5b01c31060c23ac47","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-17 17:40:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e960b6d99c99bcde424dd7bb56ce02520838032a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:40:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3a9efbbb261b859f45a736e0f9929096d2d26a02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:50:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096788ff93e34e5e93fad313fa4fea46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7202459675d64c65af0a396e50c43bcd : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f22da082ac49c78ea1fa56d02f141e : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c8689eba8204444891eeed7a8aa6c69 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5212f763db439d8b814e4bcb183ef1 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e9a2b956e54370a5bf99f933742e5d : SUCCESS in 3m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/589851f9a0474bdcadc5075543f3da33 : SUCCESS in 4m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6afe95a752da41dab6356aff498c8ba7 : SUCCESS in 4m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a58ace9ba1034493afe3fd87ac1ccc06 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"2ed6dbce0f31d92a131df54b0a63d310c35ad6b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:50:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0ad13d51df93dae3b81c1649001d5d082c7cc247","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:51:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3317fd6f84ac46df84c14e1f5429b874\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e85cc5104ef489783ee6199f7c7639d : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/659b3ac0f8ac4291875703c6d758f6fd : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6505e20baa3e240aeb9aa2c238c470eea95e0f1","revisions":{"c93e91665981eaacf38b677cdb33a4380c18bdcc":{"kind":"REWORK","_number":1,"created":"2023-01-10 19:23:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/869755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/55/869755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/1"}}},"commit":{"parents":[{"commit":"fdf10423bf29b7cfd5e88d9d06e1313ee881ab80","subject":"Make tox.ini tox 4.0.0 compatible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fdf10423bf29b7cfd5e88d9d06e1313ee881ab80"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c93e91665981eaacf38b677cdb33a4380c18bdcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c93e91665981eaacf38b677cdb33a4380c18bdcc"}]},"branch":"refs/heads/master"},"90ecfa92f030a6370658430ca3bbc6273f7ea487":{"kind":"REWORK","_number":2,"created":"2023-01-10 19:58:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/869755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/55/869755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/2"}}},"commit":{"parents":[{"commit":"fdf10423bf29b7cfd5e88d9d06e1313ee881ab80","subject":"Make tox.ini tox 4.0.0 compatible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fdf10423bf29b7cfd5e88d9d06e1313ee881ab80"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:58:50.000000000","tz":0},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/90ecfa92f030a6370658430ca3bbc6273f7ea487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/90ecfa92f030a6370658430ca3bbc6273f7ea487"}]},"branch":"refs/heads/master"},"b6505e20baa3e240aeb9aa2c238c470eea95e0f1":{"kind":"REWORK","_number":3,"created":"2023-01-10 20:56:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/869755/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/55/869755/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/55/869755/3"}}},"commit":{"parents":[{"commit":"fdf10423bf29b7cfd5e88d9d06e1313ee881ab80","subject":"Make tox.ini tox 4.0.0 compatible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/fdf10423bf29b7cfd5e88d9d06e1313ee881ab80"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 20:50:45.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nAlso, remove basepython/ignore_basepython_conflict from tox.ini. Tox 4.2.6\nstill failing with the missing interpreter for [testenv:functional{,-py38,\n-py39,-py310}]. It is due to a conflict between basepython \u003d python3\nand the version fragment in the generative target suppressed by\nignore_basepython_conflict \u003d true.\n\nThis patch removes basepython \u003d python3 assuming that developers already\nswitched for python3 in their environment as python2.7 is EOL.\n\nRef: https://review.opendev.org/c/openstack/nova/+/869545\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b6505e20baa3e240aeb9aa2c238c470eea95e0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b6505e20baa3e240aeb9aa2c238c470eea95e0f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
