)]}'
{"id":"openstack%2Fosc-placement~869768","triplet_id":"openstack%2Fosc-placement~stable%2Fxena~I4e3e5732411639054baaa9211a29e2e2c8210ac0","project":"openstack/osc-placement","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-09-29 12:10:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e3e5732411639054baaa9211a29e2e2c8210ac0","subject":"Use pypi released version of placement in functional tests","status":"MERGED","created":"2023-01-10 21:17:31.000000000","updated":"2023-09-29 12:11:31.000000000","submitted":"2023-09-29 12:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869768","meta_rev_id":"15116aefe4290e1d12f02c07acff22c08d4700fb","_number":869768,"virtual_id_number":869768,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 12:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-09-29 11:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-09-29 11:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-09-29 11:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 21:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 18:53:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-09-29 11:10:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"28e534d9ad28647394e149165dbeb16174eda77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-10 21:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ddf44881db3a085d4f5f93f6a8c68ef3a6d9bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:28:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454208e9af1a4a29966fc4c064aed71b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f37a20bef5b445ea263481eece901c6 : SUCCESS in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29750ace2544441483f743181fb61adb : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd149e08db3e41e8996b4501e3185816 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11910a6b1bfb4ca29c47eabf879d1ee5 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6900327f525f45759730806510117fac : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb6133bacdf4bb597aad32d74ff7f6a : SUCCESS in 3m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/558043c8a23046cbbdc86cfb21e47f03 : SUCCESS in 3m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/52639ff39dd14f7d8282cbda5c2a1d69 : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9319c73353a54244846dc1335ddc0ecc : SUCCESS in 3m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b69c83bb0c7493aa62553f1ad9a44fe : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74f4206eecd5682a3cc5c5bf6a007f82d7096c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 01:11:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"786a65a9aa6a21bb02bba7e3b54cdcba4b4d9ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 01:20:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e31f2a76c8641eab298899f59f0fd2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c032cae32946d2a1ede4742f7af6f1 : SUCCESS in 2m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07b7625999649d49a87113c8fbb76a6 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c6727c23a8144fab7a10f889eb0ba45 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17067f23ba70418fa65e3501a4c252cd : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abfe2aa1e284560add27e116d05b8c0 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d23c726962b444ceb910d729a0e8745c : SUCCESS in 3m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5ce4991b8df4eea9e177e7be584b743 : SUCCESS in 3m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0ad8a405b78d422fa3f22514a99cace0 : SUCCESS in 3m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5bfad9d6c044f7ea5189e6e91563a52 : SUCCESS in 3m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/309f4371d83f4ab09f6661ce754071dc : SUCCESS in 3m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c21c4ca8da87d405f80f61660cd776d8e73ed55","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-13 18:53:20.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb07fd01e40e7a9891392a9ed9bd3d41297792a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 11:10:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3838a12d4a261f82ad183a701e3d164065de5934","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-29 11:59:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e063926a93228600e12f4e2aba41a91d691b7dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:59:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f35d733cdb7cb453928bfa0dc992f339c5189403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 12:10:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec835aa1037a4c4d894418caf242ad69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fa54b622c7403ebaca50fd50db73b7 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29666f13bd2a4aae9197e22bcb8ac4de : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/605cac2a556347f890a874beb2d83b8a : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a856f0b9f647ddba1aa370fac9f0e8 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a985219e1ac47278e912af731e4dd4e : SUCCESS in 3m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3901d684fdd146c1ba43e92960bd2826 : SUCCESS in 3m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/994ceee175384fa09e1b03d144455f6e : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c694d4dee39b4fd8b135061ba46f118f : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7e2afe68864fd5258cd1ac3b27b8c467a755671d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 12:10:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"15116aefe4290e1d12f02c07acff22c08d4700fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 12:11:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/547d3f90e0d5479ba383c9e1e169fb4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7adb7d06e5b34893abb525395eedcdba : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13410360c9514b05b273412ec9edcd58 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf3a355928019216b4142d23000f802fbabee813","revisions":{"1fb8c43b7732fe5422d050a20943a8cae5da407b":{"kind":"REWORK","_number":1,"created":"2023-01-10 21:17:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/869768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/68/869768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/1"}}},"commit":{"parents":[{"commit":"486ad44e40a08c12dad2adbc6521277a55c46383","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/486ad44e40a08c12dad2adbc6521277a55c46383"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 21:17:24.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit 90ecfa92f030a6370658430ca3bbc6273f7ea487)\n(cherry picked from commit 3f2a80b68331d774947d0716151daaf0bfb94ac1)\n(cherry picked from commit 4dfbfa14c687dd4c4bbf11fa65939b84b6d68baa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1fb8c43b7732fe5422d050a20943a8cae5da407b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1fb8c43b7732fe5422d050a20943a8cae5da407b"}]},"branch":"refs/heads/stable/xena"},"bf3a355928019216b4142d23000f802fbabee813":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-11 01:11:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/869768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/68/869768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/68/869768/2"}}},"commit":{"parents":[{"commit":"486ad44e40a08c12dad2adbc6521277a55c46383","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/486ad44e40a08c12dad2adbc6521277a55c46383"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-10 19:18:02.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-03-11 01:11:12.000000000","tz":-360},"subject":"Use pypi released version of placement in functional tests","message":"Use pypi released version of placement in functional tests\n\nosc-placement functional tests are using the master version\nof placement which is failing due to the constraints mismatch\non stable branches testing.\n\n- https://zuul.opendev.org/t/openstack/build/a2cc830f28f44ed5ae0e1c5f0a99ac75\n\nDuring the discussion in Nova meeting, we did not find any reason why\nwe are using the latest placement on stable branch testing of osc-placement\n- https://meetings.opendev.org/meetings/nova/2023/nova.2023-01-10-16.00.log.html#l-47\n\nBelow is the original patch which started using the placement from master\nsource.\n- https://review.opendev.org/c/openstack/osc-placement/+/651939\n\nTo avoid the constraints mismatch on stable branch testing which can\nhappen in future also, let\u0027s use it from pypi version which will be the\nsame way as nova functional tests doing\n- https://github.com/openstack/nova/blob/26f24b7086855bc8f57059d618060cc5844dca15/tox.ini#L64\n\nDuring this backport, removing basepython changes are not applicable on stable as we have\npinned stable branches for tox\u003c4.\n\nChange-Id: I4e3e5732411639054baaa9211a29e2e2c8210ac0\n(cherry picked from commit 90ecfa92f030a6370658430ca3bbc6273f7ea487)\n(cherry picked from commit d4036554f044f548d6ba460db0114c0fd7469f7b)\n(cherry picked from commit d2c170d778c7aadccae602b9e7e95144b9d888ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bf3a355928019216b4142d23000f802fbabee813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bf3a355928019216b4142d23000f802fbabee813"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
