)]}'
{"id":"openstack%2Fosc-placement~894067","triplet_id":"openstack%2Fosc-placement~stable%2F2023.2~Ia861988ae54ea18c5f351f1f3a88587ee0c1e4ae","project":"openstack/osc-placement","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"Ia861988ae54ea18c5f351f1f3a88587ee0c1e4ae","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","status":"MERGED","created":"2023-09-07 09:37:12.000000000","updated":"2023-09-12 16:45:36.000000000","submitted":"2023-09-12 16:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894067-create-2023.2","meta_rev_id":"fcd156e2b077b2ed5a934fae9d7d0f5e6a629423","_number":894067,"virtual_id_number":894067,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 16:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-09-12 16:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-09-12 16:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 10:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 16:24:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-09-12 16:36:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b063c0bc3b98b39975ab3ca4eb543aeb9bd4c7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-07 09:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac1cb3f7b27e6cc7081d23e2e2c19d5c8af02ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:17:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/278ec883a6094ae19c11f9a43e6b9926\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93cf157350eb4d5eac9136b3bef8a8a4 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6efd76da92fa4dc895b3561f71e1c83c : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e39b63df174ab58f42b851c0bee40a : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e209de1f5b94bd793fc254835af12d9 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbed3e96ff5d45568bc68cd8b728f896 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d82b1876f7498da94a8b87c985b37f : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cddb7d421e84bb9ad4e15a610bdb624 : SUCCESS in 2m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/efb02c84a74c48089df60a5f3bb333da : ERROR Project opendev.org/openstack/placement does not have the default branch refs/heads/master in 4s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85757b145c1d45498ad29eea519c4729 : ERROR Project opendev.org/openstack/placement does not have the default branch refs/heads/master in 4s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3557dbc2ef0a4c569b728046c58c1361 : ERROR Project opendev.org/openstack/placement does not have the default branch refs/heads/master in 3s","accounts_in_message":[],"_revision_number":1},{"id":"deefe7b46b3799c398e1525420097f2a5cba89e0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-12 16:24:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6161b89ff2be6458679631c8f5cd5b3a6ed02bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:35:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c55d379c68441d9b7e4be6605b22dba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41addeff896461d958550fa651f71f3 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f74db5233ba64e0abd687049a133f6c3 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e27728777db94fa6bf63d8aaf57e1e56 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd914ff8cc3e4eccba7cca6d14089505 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7efeb7c3cfc0406f88e8ddd7fa416a39 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd5299757d8452b887274d7996fe05b : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f8b6cdd8644a5e9e66d9dc3b52267e : SUCCESS in 2m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a52067461364d19924a4f1d1fb46b56 : SUCCESS in 3m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0fc1f9754bc43bbbf274a1a2344b715 : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c1fa4dfe69841bc832c599ce678325a : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7dd5edd8ec6b75d0e259c91bc11e775cd4181507","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-12 16:36:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"caa688bffafbce1806da19619b80bb649da40b5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:36:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"847bf5ae825f41e9d1a3b1d1d3de0b006c35472c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:44:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6455f423f12a419dab1728e5c7c32519\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0d4ddab1424e61b6f1c5c00ffbf738 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/084fa9c8e79a4c6daf1bf0960f1bf80f : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/256131a89d7541f1af3cc1fe0f3e05d5 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b94317f999240908024a524b93c181d : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c07da2a18ad4baa9924bd56fd4dc6b5 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa1f3eb0882c4ebeb0dc91ca79f4c00b : SUCCESS in 3m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1038027fa7c34b408a8949ad7040639d : SUCCESS in 4m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/027fb5e74e704be8b30c6214cbbb8716 : SUCCESS in 4m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0f0a8e4991c47f28bfa7c195e880e07 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dbc216f2666e693a89cda0f42c90edc5b50a6b2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:44:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fcd156e2b077b2ed5a934fae9d7d0f5e6a629423","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f51eb84ea744824aa8ea2f72a4f0a9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b1ca84bbe04d359db43ef21bd92c41 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d5ef4777df4f76b0ef9adcc647c95d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"508c11914524eedb8e6868acedaa58da1f93445c","revisions":{"508c11914524eedb8e6868acedaa58da1f93445c":{"kind":"REWORK","_number":1,"created":"2023-09-07 09:37:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/67/894067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/67/894067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/894067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/894067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/894067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/67/894067/1"}}},"commit":{"parents":[{"commit":"46850b271df68d9cf985bc41252a43026b432e81","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/46850b271df68d9cf985bc41252a43026b432e81"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-07 09:37:11.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-07 09:37:11.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ia861988ae54ea18c5f351f1f3a88587ee0c1e4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/508c11914524eedb8e6868acedaa58da1f93445c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/508c11914524eedb8e6868acedaa58da1f93445c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
