)]}'
{"id":"openstack%2Fosc-placement~943757","triplet_id":"openstack%2Fosc-placement~stable%2F2025.1~Idb28489f3908eea1f983c8421f2534cf899f3c92","project":"openstack/osc-placement","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"Idb28489f3908eea1f983c8421f2534cf899f3c92","subject":"Update .gitreview for stable/2025.1","status":"MERGED","created":"2025-03-07 14:36:22.000000000","updated":"2025-05-06 13:39:40.000000000","submitted":"2025-05-06 13:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943757-create-2025.1","meta_rev_id":"9c2fe9b0530d48231d73e58d54dc82815617fa89","_number":943757,"virtual_id_number":943757,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 13:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2025-03-15 10:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2025-03-19 12:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2025-03-19 12:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 14:57:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-15 10:42:33.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2025-03-19 12:36:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d990732c05d56f17b56d7c1a50beb64b60d70426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-07 14:36:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7571209c0cce7bb13615c336ff1647221cf1edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:57:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ce3fbffdf54db0a17eb099c87f97cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded9a1ff65a746ce9fbac9d912b15b8f : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55ebc6b6da4540dd8886e446db1f6f60 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dee04bd274a4c36a606240b05ca51a1 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338257d47aee459f9ef3e5db2589d8e4 : SUCCESS in 4m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce1d418d1a3344d5ac5f78cea8ba110e : SUCCESS in 4m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae89c5097b664d6e932c7a04d38df0bd : SUCCESS in 3m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1bb6d54d27bd43a1b22b4139d1387366 : SUCCESS in 2m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d635da6a7ce74504b759ea75405b4f49 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"aa66f930d70ead394cecf51ed16a1d7b579ca679","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2025-03-15 10:42:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75eaed57fe7457a74303e337f5d68fc8ff22642d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-19 12:36:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06d3f7b7aa50d0232e371104c4447ed19a383dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:36:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"122f9fefb5f98251f80ae6e90eadb4513613ca2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:50:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34ac679584e14206bff2e6c5e942a6b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df03c745653a4862b0581554c5b65bab : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8b053006dbc493b99fbc29b30711a5b : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cd9031d200f4dba8cd441a60c0ec0bb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2bf18cd09b4f84bc4523d3c7176bb7 : FAILURE in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1171effcd6df47d39c310ae8656e2434 : SUCCESS in 3m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e6baff039df4a54b16309bf6728c4e1 : SUCCESS in 4m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/307ce8df2c454cbfbfd8433dbf4ec32a : SUCCESS in 3m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c4b289c4b25b4471ab8271c19f7f97ff : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d770c94f9c2bf152be98e676234494c8feec0bbd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-22 10:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33150f0578079d85d75711621cebef8674646caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 10:07:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46d2b17ef0b64df0af1441bde37f4915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23ab50049e94dd08193414d25c4357c : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec18f332f64a4896b89b786fdf748b7a : SUCCESS in 1m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a0c969b762248d0a7db8963d671b512 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e4353fbbbc47c9828e283768eecdb5 : FAILURE in 2m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/198cd296771a4d2b9c3e5c9898b757af : SUCCESS in 2m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/caa07e9802394cfe9ab331a85232566f : SUCCESS in 1m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/78917c94f5e84111b8f39c0ee17b8f9f : SUCCESS in 2m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f4869ef0d9e54b5092be65cf0ed11382 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7cf8735d703d5538538ceaa1eea4470831bf4b79","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-06 11:56:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cb743ce770558d9f01c66fd5b0e35cbfc93c7cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:07:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1de605a6dbab4d059cba6df784762c20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a34bb73b2384c0eaa1c4699f2c2c256 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d88da6df59d44e5ba5d184ef64d2e8e8 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff527d3222d54fe480aef463e9658402 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3571da65104d402598dcc5017af3f719 : FAILURE in 4m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b2e57968ac34a949d1d0cfadb6e32f8 : SUCCESS in 3m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/487e4240026d4933bd2cbc88485dcafd : SUCCESS in 3m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7377706f98d74b23b79328fefac06bd2 : SUCCESS in 3m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/89d12783aa9046d1b551951dd256c9aa : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b3a9656620a5e2d848436626d8bf9bbbe4c0b919","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-06 13:19:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df2c25199fc81faea058257249139f69617ccfc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:27:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147945452fc14372aaa5bf70345c4954\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c83f624c51a47cbabc0d5f66450d69a : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ab4ab5596c417593c11f3c6467f1f7 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba68be527964c4ca7864bc30472472a : SUCCESS in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef2bdc5641344118145be0af1407399 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec03e53ebd1246ffba73e8cd40bb4bf4 : SUCCESS in 2m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7aa7458aa2834ba09eee0aae57045bf3 : SUCCESS in 3m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/284094044ac44480a8116435cee09bb3 : SUCCESS in 2m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a696906900b241a597105a10ef6e0c48 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"022ada4449a39724eb1003f04bb84616ea025d16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:27:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8e541fc4c65d581b8a16e020f12f29e52527119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:38:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f201a5da104a92bf1161f332cdd17a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb73479a138148c4992566588649e90a : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5e7f9f2e8ea42f6a1ac4e1fd2d59a27 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0532eb3e43e34facb5b34274c171269d : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093c9a115cb44219ae4a8bc28dc0275d : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/073e4ab646db47a98164dbdc39063008 : SUCCESS in 3m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83823a46360145548a7662d407567249 : SUCCESS in 3m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e1b808bc1be84abbb11b2b34dea41505 : SUCCESS in 1m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/48e4b5f4eda640088ef59e2506159e52 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"339589ea32deea8a64f485ddac223fadf72529e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:38:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9c2fe9b0530d48231d73e58d54dc82815617fa89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:39:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e509fd3403e44ee83984f1229874263\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb58ae4acc334bda891faf15a1de7496 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5b8d568dc6eb8947c648cfc6d7b5b4ac5464d74","revisions":{"f5b8d568dc6eb8947c648cfc6d7b5b4ac5464d74":{"kind":"REWORK","_number":1,"created":"2025-03-07 14:36:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/57/943757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/943757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/943757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/943757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/943757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/943757/1"}}},"commit":{"parents":[{"commit":"454e48d05a9a2d652932491fbe19b70c37e47088","subject":"Merge \"Update bug tracker url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/454e48d05a9a2d652932491fbe19b70c37e47088"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-07 14:36:21.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-07 14:36:21.000000000","tz":0},"subject":"Update .gitreview for stable/2025.1","message":"Update .gitreview for stable/2025.1\n\nChange-Id: Idb28489f3908eea1f983c8421f2534cf899f3c92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f5b8d568dc6eb8947c648cfc6d7b5b4ac5464d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f5b8d568dc6eb8947c648cfc6d7b5b4ac5464d74"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
