)]}'
{"id":"openstack%2Fosc-placement~948867","triplet_id":"openstack%2Fosc-placement~stable%2F2025.1~I9491352d1d49a79ab659a9bc537aae2117e5e193","project":"openstack/osc-placement","branch":"stable/2025.1","topic":"osc-placement-2025.1-gate-fix","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-05-06 12:55:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9491352d1d49a79ab659a9bc537aae2117e5e193","subject":"Add bindep.txt for ubunutu 24.04 support","status":"MERGED","created":"2025-05-06 12:06:52.000000000","updated":"2025-05-06 12:56:20.000000000","submitted":"2025-05-06 12:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948867-osc-placement-2025.1-gate-fix","meta_rev_id":"b3a7f1ac22007ef5a5d21ff9744b875b34827b2c","_number":948867,"virtual_id_number":948867,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 12:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2025-05-06 12:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2025-05-06 12:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-06 12:06:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-05-06 12:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-06 12:30:30.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"ba3b6673c9fff57822877f14cb3f9dcdee4b979a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-06 12:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"197355452157604e72f4bb47a3545875448eaf45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:17:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54a0a05bd0ef439eb48afcfcf112075d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adae0690084e48db8b203a84a2ee89aa : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ace28a3bf9894510b9b234f626fb25ee : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8787f4b760af4e9e8938030b6229b17f : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be17b4e696004fb1925cac2fa540528c : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ddd02f5d6de43c1810f99171a6d5c51 : SUCCESS in 2m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0c797177ae641fab8babc5e3e4b9bee : SUCCESS in 3m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/55d6f967d05e457e8e286e40e9aaac03 : SUCCESS in 3m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3476b46e63ff4e1aae107cddad31e64c : SUCCESS in 3m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2959b3faad9749568195a52137ec84a5 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4fbabecc5b2f07767ac3ad4de7c1e75285bc404a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-06 12:20:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a20b31d5756c4fdfb5666bc74ff7229ee8017a31","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-05-06 12:30:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"045094d30ae773b28cdc421d920b5c993840af8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a245ce8bc6ac9dd11e48a27588c74f124a2640c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:55:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82634af31d54cf48f8d1b0a293873e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cb530099dc4baf90487fa4eec00dc8 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a1006ac2da54a9b8230908a68587e06 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6842c53003e54b8b8faa63e936e7359d : SUCCESS in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0420c52cc247dc81b6e90f03fb76dc : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ad6306d805b41a58f78f51c5763c94a : SUCCESS in 2m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a751ff587e29432dadc4084c56f5a861 : SUCCESS in 3m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2797682c8cc48918d8d5b2cb3284a22 : SUCCESS in 3m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/72fe62f6c81e45a3872edf6fe896b16d : SUCCESS in 3m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e53d410a2ac9452780e3dca5d5f1ddc1 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b11735e3c599792c1374a9fc047ef05980460756","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3a7f1ac22007ef5a5d21ff9744b875b34827b2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:56:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c0058d86124dbdaa04b724fe2ce5f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5159fb178b49b0b7a9f4ebe7100853 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05256e67c5cd4088b4dadf67ab14b293 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47e1ede39a0d6cd0738b93a68474c30f016009d0","revisions":{"47e1ede39a0d6cd0738b93a68474c30f016009d0":{"kind":"REWORK","_number":1,"created":"2025-05-06 12:06:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/948867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/67/948867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/948867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/948867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/67/948867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/67/948867/1"}}},"commit":{"parents":[{"commit":"454e48d05a9a2d652932491fbe19b70c37e47088","subject":"Merge \"Update bug tracker url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/454e48d05a9a2d652932491fbe19b70c37e47088"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-01 12:11:52.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-05-06 12:06:17.000000000","tz":120},"subject":"Add bindep.txt for ubunutu 24.04 support","message":"Add bindep.txt for ubunutu 24.04 support\n\nthe tox jobs now use ubuntu 24.04 by default\nand some packages that were previously installed\nimplicitly are now not installed by default.\n\nspecifically the docs jobs need pcre.h to be present\nto be able to compile some modules.\n\nChange-Id: I9491352d1d49a79ab659a9bc537aae2117e5e193\n(cherry picked from commit 35cc3479561607d167919df8250eaa89c039c740)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/47e1ede39a0d6cd0738b93a68474c30f016009d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/47e1ede39a0d6cd0738b93a68474c30f016009d0"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
