)]}'
{"id":"openstack%2Fosc-placement~959873","triplet_id":"openstack%2Fosc-placement~master~I7726c99961ef41c7cf5ec71bf28df15eee191877","project":"openstack/osc-placement","branch":"master","topic":"osc-placement-flamingo-gate-fix","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-09 16:21:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7726c99961ef41c7cf5ec71bf28df15eee191877","subject":"[CI] Drop unsupported py39 based functional job","status":"MERGED","created":"2025-09-05 15:35:37.000000000","updated":"2025-09-09 16:23:48.000000000","submitted":"2025-09-09 16:21:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959873-osc-placement-flamingo-gate-fix","meta_rev_id":"7722e05775d50797c030a03568bd48d1e5bba6bc","_number":959873,"virtual_id_number":959873,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-09 16:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-09-09 16:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-09-09 16:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-09-09 16:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 15:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-09 16:09:30.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-09-09 16:09:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"892b0b666fa3787e77e029cca3ae6eb33a2252ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-05 15:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d325acb5256be30d3664dd78fabc2f9512a0bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:50:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35878b86d7a34f6b87c0e6da154e1ff1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac7c0b5488e4c4ca48bb5ae48827f8e : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9de3937310d43ae863be985872b3b89 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e46729048dc6446985b4413923bdb7ca : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0aac481973e445cd8430dd3ce152f5e8 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26d9744789d40d785c67c695b755423 : SUCCESS in 6m 02s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5a14a779b3b24ba9a24b0b78d99e7a96 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f811a91221f841a8aa8795e5099ead1b : SUCCESS in 3m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/040973ce3bf147d9b2d4d8e4b46c7652 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd2a2d92bc2348b3800014471faf5556 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"01b340ec78b6ae865d3c076a1ac4dcc973a4ce59","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-09 16:09:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebd0a24f7c408df3bbe6ba5e7c4e211ffc01186c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-09-09 16:09:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1afbea0b6a442ce421cc16706083875aa80b006","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-09-09 16:10:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec88a5ff4d64768f2507e08a1d2ca5bbc8b035a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:10:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13cbba8490d79424c95789f2011a8a0c419f07d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:21:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292665abce424e91bc27e7255a1aea51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64be8fca2f94be9a170dfb1d0329e12 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fdd42026e51410285941fd77d96f88c : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76f67d7a12a44d10a80ba9e6a98e545c : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b83fec40e8a4b73b51ed74b7ecb7f9a : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47b37be3376b458facb882c82fc9b007 : SUCCESS in 3m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d8d2d33dd6748d983c7b68d2f272e4f : SUCCESS in 3m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9df3ec713caa43509fb0c0e244e4b04c : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4ecbbd57b92f117c0e2c885ad81f8f43362db62f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:21:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7722e05775d50797c030a03568bd48d1e5bba6bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:23:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09451c11c1034094a548d5f478eb289b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727fd6b6c8894ed3a684d1af59d6cac0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffb58eca9b90dbf0d24a379cd0cca895cba8930a","revisions":{"ffb58eca9b90dbf0d24a379cd0cca895cba8930a":{"kind":"REWORK","_number":1,"created":"2025-09-05 15:35:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/73/959873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/73/959873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/73/959873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/73/959873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/73/959873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/73/959873/1"}}},"commit":{"parents":[{"commit":"53f6f56b9cb32818a6f641ca435164472eed083e","subject":"Merge \"Drop unnecessary \u0027x\u0027 bit from doc config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/53f6f56b9cb32818a6f641ca435164472eed083e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-05 15:29:16.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-05 15:31:39.000000000","tz":120},"subject":"[CI] Drop unsupported py39 based functional job","message":"[CI] Drop unsupported py39 based functional job\n\nPython 3.9 is not supported in 2025.2 Flamingo release and with latest\nlibrary releases the py39 based functional test job started to fail, so\nlet\u0027s get rid of it.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: I7726c99961ef41c7cf5ec71bf28df15eee191877\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ffb58eca9b90dbf0d24a379cd0cca895cba8930a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ffb58eca9b90dbf0d24a379cd0cca895cba8930a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
