)]}'
{"id":"openstack%2Fosc-placement~979692","triplet_id":"openstack%2Fosc-placement~stable%2F2025.1~Id8f50cccb788380af67cb779de20f540030d1d65","project":"openstack/osc-placement","branch":"stable/2025.1","topic":"pin-setuptools-81-docs","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-03-30 16:46:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id8f50cccb788380af67cb779de20f540030d1d65","subject":"[CI][stable-only] Fix docs job","status":"MERGED","created":"2026-03-09 15:57:05.000000000","updated":"2026-03-30 16:47:50.000000000","submitted":"2026-03-30 16:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979692-pin-setuptools-81-docs","meta_rev_id":"7c2c1cc3feb809354754eef3fd4b403c28639826","_number":979692,"virtual_id_number":979692,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 16:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-03-30 16:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-03-30 16:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-03-30 16:31:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 16:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 16:30:46.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-03-30 16:31:45.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"80f2657dc6d76a8876a9e2bc016d022fad66b5d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-09 15:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91930bb1cc0c25e82a9ef3a551585a092ebe2910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:10:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23077f7e081441a898cda3dac58ed202\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1453c7ab2a6548cb8145445e5359537b : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da971459ce1459a9503a2b086cffc8e : SUCCESS in 1m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3615dcb09d914459a2965c536ada3c0f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09016b9d5d243b7b243aa5668fc3c4e : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d83d0ca8424644fa8e3837383675232c : SUCCESS in 3m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55b69bc23e794da99b252fe0d8c6a9ea : SUCCESS in 5m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/614b9b5257f846d9aa6c74b7076352b3 : SUCCESS in 3m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/80dd6a24f3514cfc87b513e4b83d93d4 : SUCCESS in 3m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b1abca4a91734c5f9d13cebcd157c10d : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7de30516d6001f86b47fce68c0ee6440f53a8fa2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-30 16:30:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9dbccb387bb149aaf507ad42e0bccc9ca8e0fbc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-03-30 16:31:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67b5c0c0093f2bbbfb27a4415e9d354dc6fb1591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:32:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6083f8f3b696af8a0dc7a671d2de70614c01fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade758a0adaf4bb4aed0fc4a32f6f9e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e3cb94be2a4cf6826da75f0a141243 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b87866e65a6426eb772ae22c8e03520 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/187b7a8a3e694ab9bf2884ba27d6467a : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2bf2e16e9a64a70894a7b9f5c2db0c0 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b29e16079e764766a828ed601d208640 : SUCCESS in 2m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cd3b8148614443fa65efcb92ed4986f : SUCCESS in 4m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9859191758b41b69a6c9adbcdee019f : SUCCESS in 4m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3f8f87e5d85f47eca5e4ac421c2d73fe : SUCCESS in 4m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/155bcfd4a1e94ae8bf2d02d5a89ab404 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"871ce259114c2217321c6207c76607a7ac70106c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:46:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c2c1cc3feb809354754eef3fd4b403c28639826","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4790c05ecf449f7898d8f2631bea179\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26bb14e895864429832349887eef992b : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4250b786fec4c01a13762256d9851d0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d263c5454dc7159bd32df5244d36f1195774755","revisions":{"8d263c5454dc7159bd32df5244d36f1195774755":{"kind":"REWORK","_number":1,"created":"2026-03-09 15:57:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/92/979692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/92/979692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/979692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/979692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/979692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/92/979692/1"}}},"commit":{"parents":[{"commit":"83fe7d255c6948fb611cc778e36ad33ea74da8e5","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/83fe7d255c6948fb611cc778e36ad33ea74da8e5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 15:56:00.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 15:56:44.000000000","tz":60},"subject":"[CI][stable-only] Fix docs job","message":"[CI][stable-only] Fix docs job\n\nsetuptools 82.0.0 release dropped pkg_resources module from its\nsource and with that docs zuul job started to fail.\npin setuptools\u003d\u003d81.0.0 for docs job, otherwise it fails with\n\u0027missing pkg_resources\u0027 error.\n\nChange-Id: Id8f50cccb788380af67cb779de20f540030d1d65\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8d263c5454dc7159bd32df5244d36f1195774755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8d263c5454dc7159bd32df5244d36f1195774755"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
