)]}'
{"id":"openstack%2Fosc-placement~998801","triplet_id":"openstack%2Fosc-placement~stable%2F2026.1~Id3b5fea00c8c75d10744eddfd27a889431fc8a6c","project":"openstack/osc-placement","branch":"stable/2026.1","topic":"do-not-merge","hashtags":[],"change_id":"Id3b5fea00c8c75d10744eddfd27a889431fc8a6c","subject":"DNM: CI health test","status":"ABANDONED","created":"2026-07-27 13:00:36.000000000","updated":"2026-07-27 13:58:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"91de334e33322d2a3ab26a1a47962c956653130c","_number":998801,"virtual_id_number":998801,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-27 13:23:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-27 13:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bafca65038a4494d2d1247e4ba79d1b68356e2fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-27 13:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baef5e7ce156409480b9543814ba554ad819161c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 13:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f015ada66a734543b1b050d8ff13ceae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8958fca1cf4b95950c720929ca9435 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0e6534dc64c4558a4129403e8eedeee : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89ccd2425f104db4b5a7f4d782c856f9 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/554133aeb41c4f099f9662790ebb36c5 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636d7a633c034abe818643f2627052e8 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbdc5546ee35483bbee032deb5ff51c3 : SUCCESS in 3m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b234265af233459d84bc79bf2ad4c5fa : SUCCESS in 4m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ed677af5f11f447e9fec3caa3c99343b : SUCCESS in 4m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e695cc3bd98d42619f61e7496b659f30 : SUCCESS in 4m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ddf21516b1a646fc958d88a062bb2e0d : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"91de334e33322d2a3ab26a1a47962c956653130c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-27 13:58:43.000000000","message":"Abandoned\n\ngate is healhty \\o/ abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cc8484cccb8077b5af29c07b31f2d51b4eaf505","revisions":{"9cc8484cccb8077b5af29c07b31f2d51b4eaf505":{"kind":"REWORK","_number":1,"created":"2026-07-27 13:00:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/01/998801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/01/998801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/01/998801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/01/998801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/01/998801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/01/998801/1"}}},"commit":{"parents":[{"commit":"6d9bfaffeffb3f1cc0116d56bde012fcfb61ff54","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6d9bfaffeffb3f1cc0116d56bde012fcfb61ff54"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 13:23:46.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-27 13:00:02.000000000","tz":120},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: Id3b5fea00c8c75d10744eddfd27a889431fc8a6c\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9cc8484cccb8077b5af29c07b31f2d51b4eaf505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9cc8484cccb8077b5af29c07b31f2d51b4eaf505"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
