)]}'
{"id":"openstack%2Fosc-placement~999813","triplet_id":"openstack%2Fosc-placement~stable%2F2026.1~I57d5e77056e643b821f1c26ae371e26c6df2f20a","project":"openstack/osc-placement","branch":"stable/2026.1","topic":"osc-placement-fix-docs-target","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-24 13:43:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57d5e77056e643b821f1c26ae371e26c6df2f20a","subject":"[tox] Use constraints for requirements in docs job","status":"MERGED","created":"2026-08-05 12:19:53.000000000","updated":"2026-08-24 13:44:37.000000000","submitted":"2026-08-24 13:43:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999813-osc-placement-fix-docs-target","meta_rev_id":"cc60289c27250d5587b1c094409e3b78754407d3","_number":999813,"virtual_id_number":999813,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 13:43:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-24 13:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-24 13:13:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 12:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 13:13:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f04625bffbefb7493c6152efea8640fba23bb38d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-05 12:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de36f0d5701a26cdeda3aa37d848c94b5ab0e4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 12:38:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/375b3c7126974be3807bf009191cac8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14909894a2be46088e99e76ffdba5fd7 : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3396fdb31e04690b01d2833f9c84820 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f5ab3c5824b416fb7995588e420234b : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba65828cbd8241689f873b657f7e19c3 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a76bef51ecc47b48065e9e9d0c4884d : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1fb240536442e391aa5351bbe5e59a : SUCCESS in 3m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a36cb2e90a14c77bfa82189fe61e043 : SUCCESS in 5m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c9cf2bbf2e4c4fef83f821cf3bd2f83d : SUCCESS in 4m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3fee14ecac7c4da781a2545f058947bb : SUCCESS in 3m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5452172a77854efe9e4c5c88216df532 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d8d3872ae4ceadda93ab4dadcaa8881690ec8e6e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 13:13:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8af623504dfea71deb165e3b329d2a30303c9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7befd9842c7531d1696491dafedb7f8a46cfc5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:43:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c93b433f6e434fa6b49fdcf8dd97ad26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d123ab6b2bd64c4d8c05d5d6ef739c09 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/326b52d1d16448dea2eb529311ab597b : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64c90c4dab9e4341a3b9d7676309f69f : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca52e09f30e469eb07272d8fee4a46b : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f4cf0073582476fb383354c08f5b6a9 : SUCCESS in 3m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2941f67042944cfcbf47bcd0e3c17cde : SUCCESS in 4m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0c6beaf05386467791ff3bfac42806c2 : SUCCESS in 4m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/41e900e6e9144533b8fe3dd88401eb31 : SUCCESS in 5m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bb219e3583534497960c42def28fdc64 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6ca98e540cc9c227f6581a029f5e0fa4791623a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:43:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc60289c27250d5587b1c094409e3b78754407d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8cd5ab164243cabcaa48877c614ef5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e71da79fce74491bd5c970a8b132d9a : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d583e5d18334ee8a155f0e1a92c97c0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b0d47b69c10a738c28b40314c56a80f1b42875e","revisions":{"9b0d47b69c10a738c28b40314c56a80f1b42875e":{"kind":"REWORK","_number":1,"created":"2026-08-05 12:19:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/13/999813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/13/999813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/13/999813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/13/999813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/13/999813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/13/999813/1"}}},"commit":{"parents":[{"commit":"6d9bfaffeffb3f1cc0116d56bde012fcfb61ff54","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6d9bfaffeffb3f1cc0116d56bde012fcfb61ff54"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-27 14:40:54.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-05 12:18:39.000000000","tz":120},"subject":"[tox] Use constraints for requirements in docs job","message":"[tox] Use constraints for requirements in docs job\n\ntox target \u0027docs\u0027 installs the dependencies (in requirements.txt)\nduring phase \u0027install_package_deps\u0027, but this does not use the added\nupper constraints. To avoid this the easiest is to preinstall them\nin \u0027install_deps\u0027 phase.\n\nConflicts:\n  tox.ini\n\nChange-Id: I57d5e77056e643b821f1c26ae371e26c6df2f20a\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n(cherry picked from commit 8045248fd9b44ec1e4c41e004d0fddad3c0969d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b0d47b69c10a738c28b40314c56a80f1b42875e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b0d47b69c10a738c28b40314c56a80f1b42875e"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
