)]}'
{"id":"openstack%2Foslo-cookiecutter~705477","triplet_id":"openstack%2Foslo-cookiecutter~master~I706a33405c71bb6dd1c358e7c0f5fea6dc96f7ec","project":"openstack/oslo-cookiecutter","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I706a33405c71bb6dd1c358e7c0f5fea6dc96f7ec","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2020-02-03 16:38:02.000000000","updated":"2020-02-04 10:04:19.000000000","submitted":"2020-02-04 10:04:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705477-1580810659324-e685ee51","meta_rev_id":"79086a9925cc5e644194d0ae877e040ad2eb5ffc","_number":705477,"virtual_id_number":705477,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 10:04:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-02-04 09:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-02-04 09:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-02-04 09:56:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 09:49:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-04 09:56:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-04 10:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40a26f70b2e8f29a6f07a4525ce18bb6e6e1f9d9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-03 16:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90617b55e7189a7aa8ea2b35796b7a1572b69c08","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-03 16:47:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d80877d098bbb877e72d790f983494a6315a21f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 16:51:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- oslo-cookiecutter-tox-cookiecutter https://zuul.opendev.org/t/openstack/build/eacf14f32f29445abdafee9948cf3252 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"743ed9498a00d2d048419481f24b8718b3e886ce","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-03 18:55:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b37d0ea3da98bd79bbe4e4e483bc844591dea2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 19:04:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- oslo-cookiecutter-tox-cookiecutter https://zuul.opendev.org/t/openstack/build/17b5d00e3c5f49e583dee9efc2a7bbef : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"16382cbf9b9e7f8c47244d48a3667851840ee67e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 09:49:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"908c2e49584bbb22f467a8407584f9c126089226","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 09:56:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49a8c8fa0eb3048e5b9f85132b0a48df459a4b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 09:56:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1daf19e68ced5c1c9cb8f386e18eff8ca4b99322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 10:04:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- oslo-cookiecutter-tox-cookiecutter https://zuul.opendev.org/t/openstack/build/340b268e06b6463485f54e8b875e7e3b : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f0b7a72d2084e8bdd98c2fe387b94694638ec8b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 10:04:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4648686a2730d303e08e72d678e1caf2f2d8c372","revisions":{"5416feddad694c39af9aa033fd5f4d15f63269e9":{"kind":"REWORK","_number":1,"created":"2020-02-03 16:38:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/77/705477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo-cookiecutter","ref":"refs/changes/77/705477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/1"}}},"commit":{"parents":[{"commit":"ac356a26a02e1af7dc7014c64b4f181b2f3db0ee","subject":"Fix openstack review link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/ac356a26a02e1af7dc7014c64b4f181b2f3db0ee"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 16:34:48.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 16:37:44.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg and remove old cruft.\n\nChange-Id: I706a33405c71bb6dd1c358e7c0f5fea6dc96f7ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/5416feddad694c39af9aa033fd5f4d15f63269e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/5416feddad694c39af9aa033fd5f4d15f63269e9"}]},"branch":"refs/heads/master"},"4648686a2730d303e08e72d678e1caf2f2d8c372":{"kind":"REWORK","_number":2,"created":"2020-02-03 18:55:12.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/77/705477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo-cookiecutter","ref":"refs/changes/77/705477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo-cookiecutter refs/changes/77/705477/2"}}},"commit":{"parents":[{"commit":"ac356a26a02e1af7dc7014c64b4f181b2f3db0ee","subject":"Fix openstack review link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/ac356a26a02e1af7dc7014c64b4f181b2f3db0ee"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 16:34:48.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 18:55:04.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg and remove old cruft.\n\nChange-Id: I706a33405c71bb6dd1c358e7c0f5fea6dc96f7ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/4648686a2730d303e08e72d678e1caf2f2d8c372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-cookiecutter/commit/4648686a2730d303e08e72d678e1caf2f2d8c372"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
