)]}'
{"id":"openstack%2Foslo-specs~725328","triplet_id":"openstack%2Foslo-specs~master~I4d14943a58d135f22039681aa5154ca3a956179f","project":"openstack/oslo-specs","branch":"master","hashtags":[],"change_id":"I4d14943a58d135f22039681aa5154ca3a956179f","subject":"Align contributing doc with oslo\u0027s policy","status":"MERGED","created":"2020-05-04 16:32:38.000000000","updated":"2020-06-23 18:04:20.000000000","submitted":"2020-06-23 18:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725328-1592935214767-47516c0e","meta_rev_id":"6fd83b2a2640ee34ffbbc739926a3d45cede3eae","_number":725328,"virtual_id_number":725328,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 18:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-22 15:38:44.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-23 17:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-23 17:54:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 15:38:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-23 17:54:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-23 18:00:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e59ecf54beab18e3534f125d77a097fdf3db24aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-04 16:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88240057f080432d0d607e86f691edbf8e3ee631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:40:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0afc748bc8764251b304155b39a0396b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"907ac45c073b50f1e5c3c4e240dfeaf127dbcb51","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:38:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fdbed45b64c07fc1383e3237d7e07fedf2bae6c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-23 17:54:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nOne nit but nothing really serious","accounts_in_message":[],"_revision_number":1},{"id":"162a5f239152aa25a2ecd31452d94427d0fd33dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 17:54:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a23d883f48209d6e26131c6483c15254e4151856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:00:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7fa3a2c70ed90f9a1f05580113b7ead26e1f892c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:00:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030ba779b8694efb82ae9562e50e6855 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6fd83b2a2640ee34ffbbc739926a3d45cede3eae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/581b6f99616546aeb05bc2365db1469e : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a882f2de4a1805b56f6f7ff9fd98ccdee8f3b00e","revisions":{"a882f2de4a1805b56f6f7ff9fd98ccdee8f3b00e":{"kind":"REWORK","_number":1,"created":"2020-05-04 16:32:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/28/725328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo-specs","ref":"refs/changes/28/725328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/28/725328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/28/725328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/28/725328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo-specs refs/changes/28/725328/1"}}},"commit":{"parents":[{"commit":"ec8c81fb0732218669bee12f45cf4ec8f5f315bc","subject":"Add contributor doc policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/ec8c81fb0732218669bee12f45cf4ec8f5f315bc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:32:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:32:20.000000000","tz":120},"subject":"Align contributing doc with oslo\u0027s policy","message":"Align contributing doc with oslo\u0027s policy\n\nRelated to:\n- https://review.opendev.org/#/c/723044/\n- https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nChange-Id: I4d14943a58d135f22039681aa5154ca3a956179f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/a882f2de4a1805b56f6f7ff9fd98ccdee8f3b00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/a882f2de4a1805b56f6f7ff9fd98ccdee8f3b00e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
