)]}'
{"id":"openstack%2Foslo-specs~744302","triplet_id":"openstack%2Foslo-specs~master~Ia96addf09cdb1774ef4c4e0850b61928bee9c49c","project":"openstack/oslo-specs","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ia96addf09cdb1774ef4c4e0850b61928bee9c49c","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 22:17:57.000000000","updated":"2020-08-02 23:30:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"070523b49bd39a03886a20da19d0eddf3daf1f61","_number":744302,"virtual_id_number":744302,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 21:17:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 21:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9c126a87cec2e373c462c1b248b79cd90fdafd5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 22:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0739b386bcaa921d5a58ada2f29ef4239edce36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 22:22:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8aaf22a92348428acbdf01d8578321 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"973c8666c4dc08557aeba27fa79f69c5151188c1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:58:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3310376af7dbb9d8b1438a4442218fccdf468d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:17:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a494dd98f7eb4c148a605dd8840552fb : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"14c4d5355f5c1ebc51552dc05e45944a08fd0109","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:30:01.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"946d6c7849db6d7bb7ae2e9b1cb5e073db3d209d","revisions":{"4b3b719dd4b85eaeb9995ccaa1283822835024c5":{"kind":"REWORK","_number":1,"created":"2020-08-01 22:17:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/744302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo-specs","ref":"refs/changes/02/744302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/1"}}},"commit":{"parents":[{"commit":"b467cae06c7fe0af0c7c7d83c528fcd290ceac9f","subject":"Merge \"Add more guidlines about how to contribute on oslo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/b467cae06c7fe0af0c7c7d83c528fcd290ceac9f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:16:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:17:58.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: Ia96addf09cdb1774ef4c4e0850b61928bee9c49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/4b3b719dd4b85eaeb9995ccaa1283822835024c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/4b3b719dd4b85eaeb9995ccaa1283822835024c5"}]},"branch":"refs/heads/master"},"946d6c7849db6d7bb7ae2e9b1cb5e073db3d209d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 19:58:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/744302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo-specs","ref":"refs/changes/02/744302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo-specs refs/changes/02/744302/2"}}},"commit":{"parents":[{"commit":"b467cae06c7fe0af0c7c7d83c528fcd290ceac9f","subject":"Merge \"Add more guidlines about how to contribute on oslo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/b467cae06c7fe0af0c7c7d83c528fcd290ceac9f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:16:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 19:57:59.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: Ia96addf09cdb1774ef4c4e0850b61928bee9c49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/946d6c7849db6d7bb7ae2e9b1cb5e073db3d209d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo-specs/commit/946d6c7849db6d7bb7ae2e9b1cb5e073db3d209d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
