)]}'
{"id":"openstack%2Foslo.cache~1003665","triplet_id":"openstack%2Foslo.cache~stable%2F2026.2~I51d3a289754334b9e26eee1f0a2c1d131e15fcaa","project":"openstack/oslo.cache","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"I51d3a289754334b9e26eee1f0a2c1d131e15fcaa","subject":"Update .gitreview for stable/2026.2","status":"MERGED","created":"2026-09-03 10:53:51.000000000","updated":"2026-09-03 13:05:59.000000000","submitted":"2026-09-03 13:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003665-create-2026.2","meta_rev_id":"dee50572ebc24dc8b4446277e72af8b1e40a3e8a","_number":1003665,"virtual_id_number":1003665,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 13:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 12:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 12:39:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 11:28:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:39:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"4c20613eee3d5fa67db19534b8c5bb6ebd55316b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-03 10:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52af78871c1019f62a2ba627dd1f3d435b3afa61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:28:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb2613baf5e74c48a63429211b110792\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62d68788dd8465caa508aa9443b372e : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74911303a7d4752a1d3998115d41c2a : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f986073ac0ac4882ae7ff7f4711b6f58 : SUCCESS in 4m 30s\n\nWarning:\n  Comments left for invalid file oslo_cache/_bmemcache_pool.py\n  Comments left for invalid file oslo_cache/_memcache_pool.py","accounts_in_message":[],"_revision_number":1},{"id":"64b10000249668553e713ed5a080bc37667924d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:39:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"2e28d5f8e2cc7e6b8b9dd73c838b68caa3f4a134","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:39:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9901a7cd53ed8387877b78ba437ebb65c9ab038b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:49:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486775af746e4a298b3eb98b6fa153f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2c0b133bff4718a9896e8c8d91cb86 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1367bb525f71499fb56e8a25d175830f : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58652ec87e244dac938324b54e400362 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"16707b3f653677cc8a549467e23df5263073abdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:56:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8d897ccfcea6cefa06a304cd4aae52ff5649e62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:05:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdabd05092ed4fbf82ba9df993fba86a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468fc274304045d5a794074f4b09173e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4461b9082a4e447da341ec3631f401a2 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47a07e6fe404364a53a86b538a07169 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"dee50572ebc24dc8b4446277e72af8b1e40a3e8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:05:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"893ee6d425d1dbef3a17c1d4175bd856da7ed697","revisions":{"1a9dde502c6b5dd97d41e9214f62fed2dd36d509":{"kind":"REWORK","_number":1,"created":"2026-09-03 10:53:51.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/65/1003665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/65/1003665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/1"}}},"commit":{"parents":[{"commit":"99a749393da401e07ea395e73ac2c75fb0744a91","subject":"Merge \"zuul: Use openstack-python3-next-jobs template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/99a749393da401e07ea395e73ac2c75fb0744a91"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:50.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:50.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: I51d3a289754334b9e26eee1f0a2c1d131e15fcaa\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/1a9dde502c6b5dd97d41e9214f62fed2dd36d509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/1a9dde502c6b5dd97d41e9214f62fed2dd36d509"}]},"branch":"refs/heads/stable/2026.2"},"893ee6d425d1dbef3a17c1d4175bd856da7ed697":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-03 12:39:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/1003665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/65/1003665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/65/1003665/2"}}},"commit":{"parents":[{"commit":"b5a1d8a064714d10cb6770e82a446140559b3387","subject":"Fix mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/b5a1d8a064714d10cb6770e82a446140559b3387"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:50.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 12:39:03.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: I51d3a289754334b9e26eee1f0a2c1d131e15fcaa\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/893ee6d425d1dbef3a17c1d4175bd856da7ed697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/893ee6d425d1dbef3a17c1d4175bd856da7ed697"}]},"branch":"refs/heads/stable/2026.2","description":"Rebase","conflicts":{"base":"99a749393da401e07ea395e73ac2c75fb0744a91","ours":"1a9dde502c6b5dd97d41e9214f62fed2dd36d509","theirs":"b5a1d8a064714d10cb6770e82a446140559b3387","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
