)]}'
{"id":"openstack%2Foslo.cache~1003666","triplet_id":"openstack%2Foslo.cache~stable%2F2026.2~Id8632c14096c3591fe903138fb9076d9abb3c9e9","project":"openstack/oslo.cache","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"Id8632c14096c3591fe903138fb9076d9abb3c9e9","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-03 10:53:52.000000000","updated":"2026-09-03 13:08:47.000000000","submitted":"2026-09-03 13:06:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003666-create-2026.2","meta_rev_id":"794ddddc5306752b2c6f4998850ac2bc8a72a080","_number":1003666,"virtual_id_number":1003666,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 13:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 12:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 12:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 11:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:39:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"9cdce0030f5d3df4b79f64577f21fd0d7e97f410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-03 10:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb68eb07f672361e9cab986e28c480eb3a168e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:30:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31369fb32a70443ca919bf95280f5330\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b028a014e274aeebdb5d5b61c7472ca : FAILURE in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78ba96b639d54acc9479c9ae201df1ec : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83465fe7d2864ea68a1d49c7ea9aa05d : SUCCESS in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f70a4fe0bf6f46059f5058a35223c13f : SUCCESS in 7m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3abe2915fb5d4e1fa82721f3b6e84747 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1060354394414f978cfc757b2fbea6 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5cef75fd9f243309190b7a4fc06e22b : SUCCESS in 3m 41s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/1f6b43720ecf435481da6e9c7df7df16 : SUCCESS in 5m 30s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/8564202e6bb641a39eef2f2477510a3f : SUCCESS in 4m 46s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/35af6745090c4f0b9b86f7b275061cb0 : SUCCESS in 2m 50s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/5cdf980fca5f4feeb1ce747a3e075da6 : SUCCESS in 4m 27s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/bbb93cd663494f32ab69dda67586c60f : SUCCESS in 4m 17s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/2933ef07db5d4bae94ccaa905b5a2d99 : SUCCESS in 2m 52s\n\nWarning:\n  Comments left for invalid file oslo_cache/_bmemcache_pool.py\n  Comments left for invalid file oslo_cache/_memcache_pool.py","accounts_in_message":[],"_revision_number":1},{"id":"2295032c0fa65c3add3ec516729fd0e161633c16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:39:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"bf547e91b7e3cbd7455c6782a71701a6f39b052e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:39:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e416f3ce77a548303e266a3c8956f0c254e3d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:50:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62f22b6a4a694c4ba4663a9160cba6d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f26baa984884ccabb9b09a8c386b731 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d77cd556e62d4fb99473915ecd580f84 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ac04786a4614eed82fc98bbe9b4e167 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1a1abdbcbce4128a986624d0301ec2e : SUCCESS in 4m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9d90cbff2a0e45cba8fbb182239f7333 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce5fc6c15d6426491247afb6b80ea4c : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6e8ef72feab47d89e357722f0cd83ca : SUCCESS in 3m 25s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/144d4430773348b4aa27e9750faf61cc : SUCCESS in 2m 27s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/01f7c6ba728443888d3ea6115d86a2a5 : SUCCESS in 4m 35s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/9a3f1c7d800f43b2b5dd6ac8e4393285 : SUCCESS in 2m 53s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/017bdc98456d4d10b8362a12eb15492b : SUCCESS in 4m 42s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/da8e124438a84a1bbeb6ab2042626393 : SUCCESS in 3m 37s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/8b3b15bd35504b448864cf83bcafd7b1 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0cee5a189b7d7fa39cf9cc1862b175fea1d7dd5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:56:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e500bb1f6ce09b5267744e3aa460912331c268a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:06:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409827697bd54d51a2561845a92eb14c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc67c2c12d1444b8e95c9f19e03c87e : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7b1eaa6d616451ca86e5f03da26e2eb : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fe14d1ef6a0486caef433d8768f82a3 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12d90e707b7e4896970cde3800be987f : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a270f854446b486888df55e9174c66c1 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d634c96875f347239b4acd39d2261e10 : SUCCESS in 3m 21s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/3365aac3abbe468dbb8973a8c4ed525a : SUCCESS in 3m 44s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/deab93a73f5043ef91d25a1eefdce497 : SUCCESS in 3m 41s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/a1461f4f38814d938088732c224c7800 : SUCCESS in 3m 35s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/112472eb74bc436bb671e95aaea0f19a : SUCCESS in 3m 39s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/ea8f4b9623b74fe3bc8ab20900f862ef : SUCCESS in 3m 59s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/38a44a302f61445d8e53a4cc0bb7a693 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f06386eabf3695fd05552e994b9e226eaf8fbd6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:06:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"794ddddc5306752b2c6f4998850ac2bc8a72a080","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:08:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476cfa38252f4af4bc2f7b2a9b5d61b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f892d4b1434ac2a41a1f6da4e36660 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ed6c8455f9436fa6930809368a201f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc96999a99e4a6b5d042e606bbbd7fa88ece7694","revisions":{"fddff61dbb24aa4b9d487ab228fe8511f24ac704":{"kind":"REWORK","_number":1,"created":"2026-09-03 10:53:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/66/1003666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/66/1003666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/1"}}},"commit":{"parents":[{"commit":"1a9dde502c6b5dd97d41e9214f62fed2dd36d509","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/1a9dde502c6b5dd97d41e9214f62fed2dd36d509"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:51.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:51.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Id8632c14096c3591fe903138fb9076d9abb3c9e9\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/fddff61dbb24aa4b9d487ab228fe8511f24ac704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/fddff61dbb24aa4b9d487ab228fe8511f24ac704"}]},"branch":"refs/heads/stable/2026.2"},"fc96999a99e4a6b5d042e606bbbd7fa88ece7694":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-03 12:39:10.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/1003666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/66/1003666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/66/1003666/2"}}},"commit":{"parents":[{"commit":"893ee6d425d1dbef3a17c1d4175bd856da7ed697","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/893ee6d425d1dbef3a17c1d4175bd856da7ed697"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 10:53:51.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 12:39:10.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Id8632c14096c3591fe903138fb9076d9abb3c9e9\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/fc96999a99e4a6b5d042e606bbbd7fa88ece7694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/fc96999a99e4a6b5d042e606bbbd7fa88ece7694"}]},"branch":"refs/heads/stable/2026.2","description":"Rebase","conflicts":{"base":"1a9dde502c6b5dd97d41e9214f62fed2dd36d509","ours":"fddff61dbb24aa4b9d487ab228fe8511f24ac704","theirs":"893ee6d425d1dbef3a17c1d4175bd856da7ed697","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
