)]}'
{"id":"openstack%2Foslo.cache~1003760","triplet_id":"openstack%2Foslo.cache~stable%2F2026.2~I5809e3e0523c53bd916d6f1eeb93b62110af0460","project":"openstack/oslo.cache","branch":"stable/2026.2","hashtags":[],"change_id":"I5809e3e0523c53bd916d6f1eeb93b62110af0460","subject":"Fix mypy errors","status":"MERGED","created":"2026-09-03 12:38:43.000000000","updated":"2026-09-03 13:07:08.000000000","submitted":"2026-09-03 13:05:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003760","meta_rev_id":"a1ea999f0d70d76dbd17c30730ac5995acab2e8b","_number":1003760,"virtual_id_number":1003760,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 13:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 12:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 12:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 12:38:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-09-03 12:55:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23e087ad1c7c5364e60a33d91817ae176170e524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1540ab14291359320090801eff0f5f03fcad1512","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:39:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ca78739682216de3f0db12aaf097791ac555eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:55:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d8f4174a8c41e6a3a99e1199a840fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab1efded5064bc3bb66100d3d1ad9dc : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/286f5077fd9e4731bdab591bdbc18746 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce4cb22505b045a2b77e3ddd43d5684f : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed1c328b4d18415c94da1bc6446f19f3 : SUCCESS in 3m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d4654fa4533142eeb392e404cb58116a : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4a344ce95841018c1de46c5c7ff5fe : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b9e67c46c146918176141a020721d1 : SUCCESS in 4m 45s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/cdb356cf158b454dbcf268f55ad90f51 : SUCCESS in 2m 36s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/5d14f8916d8f48db94ee295ed6f7e865 : SUCCESS in 4m 22s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/6b88d17f160448b7a509e6a1c388e120 : SUCCESS in 3m 51s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/c0d7f70efe5a48bd89249a9689ea7b21 : SUCCESS in 4m 27s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/dd2608047cb04b6e9dd6f1eebbba56be : SUCCESS in 4m 10s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/fa79f4f12cc34b4bb743c00e50ae83fd : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d6db724d193adebd58708751bb3e45245b8a2b96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:56:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83d5d8f863a2def55b6f80083bb26a8ddeff9f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:05:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8457705b97124ed1a04eb4bd269352c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/162f8e33821b409a87f442158ccff8dd : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86ee1a5fbd9942dc80fc9b7d30b65739 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfe6dddd970c4d309d756796e0c2239a : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88777bec0a014337b0f7b116b3b53727 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cd6cc7316b49ed88d661ff4f61c566 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6027513d9b0949bb91898ccb2e9c7fbb : SUCCESS in 3m 17s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/8e380e194fec480397771cfb7e4b32f1 : SUCCESS in 3m 43s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/ea9d01cbb60540dfbc45cbfa2b6ffc0e : SUCCESS in 4m 45s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/ec41fb1a844249469e1ceca8a8a502cf : SUCCESS in 3m 22s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/6c7c6189dc6f43cb8972125e2c655b4b : SUCCESS in 3m 38s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/27e8f67408a64ab6a6e49796d649dccf : SUCCESS in 4m 15s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/5d2d1653aefd452382181bd20ac76953 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f407431b3b80c8a5d5d55b426fb98ba1f51b2288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:05:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1ea999f0d70d76dbd17c30730ac5995acab2e8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:07:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7dbf58e69d443f09a675d57798c324f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0430b699dc44e31ba50bb5c9d5fe37b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5a1d8a064714d10cb6770e82a446140559b3387","revisions":{"b5a1d8a064714d10cb6770e82a446140559b3387":{"kind":"REWORK","_number":1,"created":"2026-09-03 12:38:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/1003760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/60/1003760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/60/1003760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/60/1003760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/60/1003760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/60/1003760/1"}}},"commit":{"parents":[{"commit":"99a749393da401e07ea395e73ac2c75fb0744a91","subject":"Merge \"zuul: Use openstack-python3-next-jobs template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/99a749393da401e07ea395e73ac2c75fb0744a91"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 02:24:08.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 12:38:31.000000000","tz":60},"subject":"Fix mypy errors","message":"Fix mypy errors\n\nFix the following errors detected.\n\noslo_cache/_memcache_pool.py:280:16: error: Incompatible return value\ntype (got \"type[_MemcacheClient]\", expected \"_MemcacheClient\")\n[return-value]\n\noslo_cache/_bmemcache_pool.py:91:16: error: Incompatible return value\ntype (got \"type[_BMemcacheClient]\", expected \"_BMemcacheClient\")\n[return-value]\n\nChange-Id: I5809e3e0523c53bd916d6f1eeb93b62110af0460\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit a14d55b2212f404014ce16b3fad605de3dde81c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/b5a1d8a064714d10cb6770e82a446140559b3387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/b5a1d8a064714d10cb6770e82a446140559b3387"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
