)]}'
{"id":"openstack%2Foslo.cache~1004893","triplet_id":"openstack%2Foslo.cache~master~I7fb369ac4c3e73e8922c269af40eaf1258905a0c","project":"openstack/oslo.cache","branch":"master","hashtags":[],"change_id":"I7fb369ac4c3e73e8922c269af40eaf1258905a0c","subject":"Ensure recursive detection of package files","status":"MERGED","created":"2026-09-09 16:38:34.000000000","updated":"2026-09-14 04:47:08.000000000","submitted":"2026-09-14 04:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004893","meta_rev_id":"70dd537b57c5e7427dfac3cf5277f3bff561e3ff","_number":1004893,"virtual_id_number":1004893,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 04:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-14 03:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-14 03:50:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 18:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c962584d6d941295b0f5999384dd7bf2eb92fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 16:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98c3846583881ab292b1d674afb0ef0a14217dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:17:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aa8c0d07c804943a5849fa280b26343\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1b523f4b65541cd8784054c876743ab : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b756ea6e26934f6daa6e0d744fe0e8cd : SUCCESS in 4m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d09575e86a9e46c89b87e03c574dcce2 : SUCCESS in 1h 31m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee3d18edab3423a8a196e10b9338928 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d2fb40f68654caaa16a920dd1e44cf7 : SUCCESS in 6m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8f6a626dfb342f89190baab8d390e41 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9da3db92f11a44c8b1d43056b0d2ce54 : SUCCESS in 5m 04s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c920ed13007948adb82bb897db0e35c6 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb8d33983f943ebaaae0307786b1814 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/024be85db54749a7984691442fa6223d : SUCCESS in 3m 31s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/720ce83ad31c4badb920d5982a9b2139 : SUCCESS in 7m 51s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/b3d6650168304fe696df10ad752beab1 : SUCCESS in 6m 44s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/295cbee5a00042f39148c2f4f5d6fb55 : SUCCESS in 4m 41s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/c3d0954bf5414f4d9b5d7804ef97e881 : SUCCESS in 4m 05s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/192d2b7c232a44498a2f8e8a909dc61e : SUCCESS in 4m 15s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/b684d879c3854d6c80890eaf1722af58 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"09de79e2b2512996662f6ef70043f6319248e435","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 03:50:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a9dc235a64fea98fa829d51c99cfd258e589103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 03:51:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da855d54e9bff0bbc173d8a5eedbf66b4a0494c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:46:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3167a3521524ce6ac4a1c2e903fa2a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc407fe0004945be801159a67df202cb : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/421069c499dc4fc5bb3c87fab946757a : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23b7cb3a08454c19a95afeab03da774a : SUCCESS in 54m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87eea61fb3947b8a74fe28c900d12f8 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7db8a006907f47d0a4b806bec1a9024d : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd41ac53e88f4ba88bc773162775a066 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c48074e680194fa9921822af24ab0360 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db877e11920348d1875fb02f839524e5 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9936fcc64554ec480ce98d253402356 : SUCCESS in 3m 38s\n- oslo.cache-functional-etcd3gw https://zuul.opendev.org/t/openstack/build/c6a74159d9874e86be397ca3e863f824 : SUCCESS in 4m 14s\n- oslo.cache-functional-dogpile.cache.bmemcached https://zuul.opendev.org/t/openstack/build/fcfa76ea53d64f61bff7eb89776ddcad : SUCCESS in 3m 50s\n- oslo.cache-functional-dogpile.cache.pymemcache https://zuul.opendev.org/t/openstack/build/1b0c6d569b7143768506797287dc39cd : SUCCESS in 7m 06s\n- oslo.cache-functional-memcache_pool https://zuul.opendev.org/t/openstack/build/e4102da4941f4517a4404e94391c06cb : SUCCESS in 3m 21s\n- oslo.cache-functional-dogpile.cache.redis https://zuul.opendev.org/t/openstack/build/f9df811e82044bf388e807b1b888873e : SUCCESS in 6m 33s\n- oslo.cache-functional-dogpile.cache.redis_sentinel https://zuul.opendev.org/t/openstack/build/b0ca47090cdd488ab080eae36c86e4e9 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2de2ddbe6e50bf05adc9aad355a3f097f08ddcc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:46:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"70dd537b57c5e7427dfac3cf5277f3bff561e3ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:47:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff5b23ec30c46428a0bebba759e5b17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30f45fa427d42a191153509fa383b7f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3118f21e7da5511eebcf93de41a02ff2a696bf9","revisions":{"f3118f21e7da5511eebcf93de41a02ff2a696bf9":{"kind":"REWORK","_number":1,"created":"2026-09-09 16:38:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/1004893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/93/1004893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/93/1004893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/93/1004893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/93/1004893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/93/1004893/1"}}},"commit":{"parents":[{"commit":"e92bc7beafefa30d0775dc93b963c5d91d714c5e","subject":"Remove duplicated optional dependencies from test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/e92bc7beafefa30d0775dc93b963c5d91d714c5e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 16:37:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 16:38:17.000000000","tz":540},"subject":"Ensure recursive detection of package files","message":"Ensure recursive detection of package files\n\nChange-Id: I7fb369ac4c3e73e8922c269af40eaf1258905a0c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/f3118f21e7da5511eebcf93de41a02ff2a696bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/f3118f21e7da5511eebcf93de41a02ff2a696bf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
