)]}'
{"id":"openstack%2Foslo.cache~423038","triplet_id":"openstack%2Foslo.cache~master~I84d4dde64349a45cc5c38163da107a746faddc77","project":"openstack/oslo.cache","branch":"master","topic":"Switch_to_oslo","hashtags":[],"change_id":"I84d4dde64349a45cc5c38163da107a746faddc77","subject":"Switch to oslo_log","status":"MERGED","created":"2017-01-20 05:19:09.000000000","updated":"2017-03-08 17:04:08.000000000","submitted":"2017-03-08 17:04:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88ce0ed2d7292585db809bbf644ddb1ac8f595b7","_number":423038,"virtual_id_number":423038,"owner":{"_account_id":22020,"name":"avnish","email":"avnish.pal@nectechnologies.in","username":"avnish"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2017-03-08 17:04:08.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2017-03-08 15:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-03-02 13:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2017-03-08 15:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-02 13:23:48.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2017-03-08 15:45:40.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2017-03-08 17:04:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b18da878a382ecbc3aa675f6b5118e2cea3923f6","author":{"_account_id":22020,"name":"avnish","email":"avnish.pal@nectechnologies.in","username":"avnish"},"date":"2017-01-20 05:19:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c3e614b0f4676e749c19569df699d6976db861","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 06:53:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.cache-docs-ubuntu-xenial http://docs-draft.openstack.org/38/423038/1/check/gate-oslo.cache-docs-ubuntu-xenial/95936ab//doc/build/html/ : SUCCESS in 2m 01s\n- gate-oslo.cache-pep8-ubuntu-xenial http://logs.openstack.org/38/423038/1/check/gate-oslo.cache-pep8-ubuntu-xenial/f35f984/ : SUCCESS in 3m 10s\n- gate-oslo.cache-python27-ubuntu-xenial http://logs.openstack.org/38/423038/1/check/gate-oslo.cache-python27-ubuntu-xenial/92295e6/ : SUCCESS in 2m 05s\n- gate-oslo.cache-python35 http://logs.openstack.org/38/423038/1/check/gate-oslo.cache-python35/e53e59a/ : SUCCESS in 1m 53s\n- gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial http://logs.openstack.org/38/423038/1/check/gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial/96c89d1/ : SUCCESS in 1h 33m 58s\n- gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial-newton http://logs.openstack.org/38/423038/1/check/gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial-newton/c78ae3b/ : FAILURE in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbf443a670aa1d130c48ecfb1c9e3d3607587e34","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-03-02 13:23:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25e45ff128b1c958b20d24c17974afbc7b284b56","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-03-08 15:45:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42e3b3ca97767da80824f42158a29ab4423fe6a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-08 15:45:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a2fbfa0790867d78116162a35da1cfa2a2eda202","date":"2017-03-08 17:04:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"46ceab9a3aebdd76aebf12099b7f89a348edbede","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-08 17:04:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.cache-docs-ubuntu-xenial http://docs-draft.openstack.org/38/423038/1/gate/gate-oslo.cache-docs-ubuntu-xenial/d6703a3//doc/build/html/ : SUCCESS in 1m 54s\n- gate-oslo.cache-pep8-ubuntu-xenial http://logs.openstack.org/38/423038/1/gate/gate-oslo.cache-pep8-ubuntu-xenial/f7c13e4/ : SUCCESS in 3m 05s\n- gate-oslo.cache-python27-ubuntu-xenial http://logs.openstack.org/38/423038/1/gate/gate-oslo.cache-python27-ubuntu-xenial/068c1bb/ : SUCCESS in 3m 46s\n- gate-oslo.cache-python35 http://logs.openstack.org/38/423038/1/gate/gate-oslo.cache-python35/f9a8ace/ : SUCCESS in 3m 02s\n- gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial http://logs.openstack.org/38/423038/1/gate/gate-tempest-dsvm-neutron-src-oslo.cache-ubuntu-xenial/7e5775f/ : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61dd30d36d14693a572e4ebee4fb677b9a4c44bf","revisions":{"61dd30d36d14693a572e4ebee4fb677b9a4c44bf":{"kind":"REWORK","_number":1,"created":"2017-01-20 05:19:09.000000000","uploader":{"_account_id":22020,"name":"avnish","email":"avnish.pal@nectechnologies.in","username":"avnish"},"ref":"refs/changes/38/423038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/38/423038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/38/423038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/38/423038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/38/423038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/38/423038/1"}}},"commit":{"parents":[{"commit":"eeb787db7439f057eb48fa80cb0e28752e7ea8a8","subject":"Remove references to Python 3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/eeb787db7439f057eb48fa80cb0e28752e7ea8a8"}]}],"author":{"name":"avnish","email":"avnish.pal@nectechnologies.in","date":"2017-01-20 05:15:48.000000000","tz":330},"committer":{"name":"avnish","email":"avnish.pal@nectechnologies.in","date":"2017-01-20 05:16:57.000000000","tz":330},"subject":"Switch to oslo_log","message":"Switch to oslo_log\n\nOslo cache is currently using python logging module to\nimplement logging functionality.\nThis patch made a switch to oslo.log(logging for openstack projects)\n\nReference:-\nhttp://docs.openstack.org/developer/oslo.log\n\nChange-Id: I84d4dde64349a45cc5c38163da107a746faddc77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/61dd30d36d14693a572e4ebee4fb677b9a4c44bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/61dd30d36d14693a572e4ebee4fb677b9a4c44bf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eeb787db7439f057eb48fa80cb0e28752e7ea8a8","is_merged_in_target_branch":true,"change_id":"I3986b19a28f8ac918359abadfde2a8c3ba19f639","change_number":418002,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
