)]}'
{"id":"openstack%2Foslo.cache~657081","triplet_id":"openstack%2Foslo.cache~master~I0f6ca974dc3055aad28638ef4b087a19935b4c29","project":"openstack/oslo.cache","branch":"master","topic":"trivial","hashtags":[],"change_id":"I0f6ca974dc3055aad28638ef4b087a19935b4c29","subject":"Fix lower-constraints","status":"MERGED","created":"2019-05-03 23:06:49.000000000","updated":"2019-05-04 16:05:27.000000000","submitted":"2019-05-04 16:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657081-1556985927133-11ac1cc1","meta_rev_id":"b02305a278af720b6f9a068a160982a943c7838a","_number":657081,"virtual_id_number":657081,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-05-04 07:26:53.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-04 16:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-04 14:07:49.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-04 14:07:49.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 07:26:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-04 14:07:49.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-04 16:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc691609538bba8d06a6de59d69b0dfdbf4c9788","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 23:06:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40ef7860a050ab12517e60f5386454a8f4b579e5","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-03 23:39:03.000000000","message":"Patch Set 1:\n\nIt looks like this is failing requirements, but at least lower-constraints is passing. I guess we need to actually bump the minimum in this repo too.","accounts_in_message":[],"_revision_number":1},{"id":"4a40c0656587fd79030eb38881a67646c41dfe0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:50:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/81/657081/1/check/requirements-check/8a474e9/ : FAILURE in 3m 39s\n- tempest-full http://logs.openstack.org/81/657081/1/check/tempest-full/ea4f18a/ : SUCCESS in 1h 38m 32s\n- tempest-full-py3 http://logs.openstack.org/81/657081/1/check/tempest-full-py3/c7e4886/ : SUCCESS in 1h 34m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/657081/1/check/openstack-tox-lower-constraints/7d8bb70/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/81/657081/1/check/openstack-tox-pep8/40817bf/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/81/657081/1/check/openstack-tox-py27/f411716/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/81/657081/1/check/openstack-tox-py36/456b6a9/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/81/657081/1/check/openstack-tox-py37/ac0e720/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/81/657081/1/check/openstack-tox-docs/ea9183c/html/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e2ca7e7718789ab8e6a17bc451b7a72644efce22","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-04 04:59:36.000000000","message":"Patch Set 1: Code-Review-1\n\nyou need to update requirements.txt as well","accounts_in_message":[],"_revision_number":1},{"id":"b5d78a5955a76f2ca1d5ff95a7d05adf881c3cc5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-04 05:03:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2a9e4bf1e827e0b25f92ec788210506afc2f10f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-04 05:03:36.000000000","message":"Patch Set 1:\n\nI\u0027ve just fixed the requirements.txt file...","accounts_in_message":[],"_revision_number":1},{"id":"5dbc134353e6c319f2d90c0ccf5888d561a5dbb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 07:09:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/81/657081/2/check/requirements-check/3d7f0fe/ : SUCCESS in 5m 11s\n- tempest-full http://logs.openstack.org/81/657081/2/check/tempest-full/97e4f93/ : TIMED_OUT in 2h 04m 36s\n- tempest-full-py3 http://logs.openstack.org/81/657081/2/check/tempest-full-py3/cb39e7b/ : SUCCESS in 1h 30m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/657081/2/check/openstack-tox-lower-constraints/2a26f6a/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/81/657081/2/check/openstack-tox-pep8/99976c7/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/81/657081/2/check/openstack-tox-py27/8d4b159/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/81/657081/2/check/openstack-tox-py36/9ceb878/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/81/657081/2/check/openstack-tox-py37/6a4b319/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/81/657081/2/check/openstack-tox-docs/4981e41/html/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fdb96e9f0f6a9b819b94af3374331fbd84443a19","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-04 07:26:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb8c0d31bdaaabff3b579583efa1894ac99c3d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 09:18:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/81/657081/2/check/requirements-check/26eea72/ : SUCCESS in 3m 36s\n- tempest-full http://logs.openstack.org/81/657081/2/check/tempest-full/ae457fe/ : SUCCESS in 1h 45m 48s\n- tempest-full-py3 http://logs.openstack.org/81/657081/2/check/tempest-full-py3/6221acc/ : SUCCESS in 1h 40m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/657081/2/check/openstack-tox-lower-constraints/b68ee5a/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/657081/2/check/openstack-tox-pep8/e28b725/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/81/657081/2/check/openstack-tox-py27/99a6c38/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/81/657081/2/check/openstack-tox-py36/ce29b9e/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/81/657081/2/check/openstack-tox-py37/93105c4/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/81/657081/2/check/openstack-tox-docs/1a50419/html/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"65149e9e27dafd51102472e571ae08008d85673f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-04 14:07:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05f4e051abd9bafbbe914802584acd7763ea725e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 14:08:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e148d594dc3a21098fd03a7e2c072452d7d25bd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:05:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/81/657081/2/gate/requirements-check/fdb134b/ : SUCCESS in 3m 33s\n- tempest-full http://logs.openstack.org/81/657081/2/gate/tempest-full/98afd54/ : SUCCESS in 1h 52m 05s\n- tempest-full-py3 http://logs.openstack.org/81/657081/2/gate/tempest-full-py3/7ac2327/ : SUCCESS in 1h 46m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/657081/2/gate/openstack-tox-lower-constraints/9dcd616/ : SUCCESS in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/81/657081/2/gate/openstack-tox-pep8/c30e6df/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/81/657081/2/gate/openstack-tox-py27/3299f13/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/81/657081/2/gate/openstack-tox-py36/81428d8/ : SUCCESS in 6m 46s\n- openstack-tox-py37 http://logs.openstack.org/81/657081/2/gate/openstack-tox-py37/51b113c/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/81/657081/2/gate/openstack-tox-docs/1d1dcd3/html/ : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7746641c1c7f470f03059fb033a0a75caf04f46a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8625175022068a71ae7eefd3dbcfcd4eaa1b7b2","revisions":{"8ceb02eeba51adf3d54f2dc501b99e65a48d5cca":{"kind":"REWORK","_number":1,"created":"2019-05-03 23:06:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/657081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/81/657081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/1"}}},"commit":{"parents":[{"commit":"dd394012f447dc8af78eb4823db4f1070b3f6aef","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/dd394012f447dc8af78eb4823db4f1070b3f6aef"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-03 23:05:06.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-03 23:06:20.000000000","tz":-360},"subject":"Fix lower-constraints","message":"Fix lower-constraints\n\nWe require six \u003e\u003d 1.11.0 due to our oslo.log dependency. Fix this. We\nwon\u0027t actually see this fix until we remove the use of \u0027tox_install.sh\u0027\nin I48f31084d595c44114391e046e5fe418c1fb787a.\n\nChange-Id: I0f6ca974dc3055aad28638ef4b087a19935b4c29\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/8ceb02eeba51adf3d54f2dc501b99e65a48d5cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/8ceb02eeba51adf3d54f2dc501b99e65a48d5cca"}]},"branch":"refs/heads/master"},"c8625175022068a71ae7eefd3dbcfcd4eaa1b7b2":{"kind":"REWORK","_number":2,"created":"2019-05-04 05:03:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/657081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/81/657081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/81/657081/2"}}},"commit":{"parents":[{"commit":"dd394012f447dc8af78eb4823db4f1070b3f6aef","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/dd394012f447dc8af78eb4823db4f1070b3f6aef"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-03 23:05:06.000000000","tz":-360},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-04 05:01:53.000000000","tz":120},"subject":"Fix lower-constraints","message":"Fix lower-constraints\n\nWe require six \u003e\u003d 1.11.0 due to our oslo.log dependency. Fix this. We\nwon\u0027t actually see this fix until we remove the use of \u0027tox_install.sh\u0027\nin I48f31084d595c44114391e046e5fe418c1fb787a.\n\nChange-Id: I0f6ca974dc3055aad28638ef4b087a19935b4c29\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/c8625175022068a71ae7eefd3dbcfcd4eaa1b7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/c8625175022068a71ae7eefd3dbcfcd4eaa1b7b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
