)]}'
{"id":"openstack%2Foslo.cache~773904","triplet_id":"openstack%2Foslo.cache~stable%2Fussuri~Ief83bd490826d90eb5239ebb71394aa516f033b7","project":"openstack/oslo.cache","branch":"stable/ussuri","topic":"memcached-tls","hashtags":[],"change_id":"Ief83bd490826d90eb5239ebb71394aa516f033b7","subject":"Add TLS options to oslo.cache","status":"MERGED","created":"2021-02-03 14:21:03.000000000","updated":"2021-02-10 16:40:39.000000000","submitted":"2021-02-10 16:38:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773904-memcached-tls","cherry_pick_of_change":725405,"cherry_pick_of_patch_set":4,"meta_rev_id":"fd22121c00c7dcd221243c0a946c0a1d289cc0f9","_number":773904,"virtual_id_number":773904,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 16:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-02-10 16:40:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-02-10 15:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-10 16:40:39.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-02-10 15:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 14:44:21.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-02-03 14:44:21.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2021-02-03 14:44:21.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-03 15:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 16:27:44.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2021-02-08 16:27:48.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-02-09 15:22:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"5b6671652d520ca1d307f8ffd7672dc50531d53b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-03 14:21:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"37c3ae4199040f25dbc67e4c18b3e84a91ab30e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:55:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e431269014a47a1be5fd43cf75ce3de : SUCCESS in 1h 28m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88740e456d824824b0e1d1cc888b7baf : FAILURE in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd91069cdb774f14ad56d15c3e39167c : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0149d00a204494cb178e1a6b788ce03 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6229fd2be5ee4b0d8d44b2cb809f7b5e : SUCCESS in 3m 28s\n- oslo.cache-tempest-full-py3-memcached-pool https://zuul.opendev.org/t/openstack/build/a2aae60fcab244a981eb6af56e05a3b6 : SUCCESS in 1h 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9233cf4f754537e6329946ce6fe601fd4298233d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-03 16:04:06.000000000","message":"Abandoned\n\nnot backportable","accounts_in_message":[],"_revision_number":1},{"id":"61543ccb55d05942ba2326bcc72704228ebee173","tag":"autogenerated:gerrit:restore","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-05 10:39:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"105d194d44cad29a2156032680aafeb46ae54638","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-05 10:56:49.000000000","message":"Topic tls-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"8aa8fc6ab298a70808efea72c1372e3d840685c4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-05 10:57:31.000000000","message":"Topic set to memcached-tls-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"5b9fe356a9eb37f5d684eeb7de3542cfc09e6525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-05 11:25:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fd0562daacadc58e634e21018c6068dae5684a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:51:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73b1d51aa6d94e189a7781fd5743b3ad : SUCCESS in 1h 23m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48da92f73a954ebeb9737461ece18fb1 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ece17fc42214005bcaa66a49872cf63 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cb82cb7d3442b8a51ddc0c029fce50 : SUCCESS in 3m 15s\n- oslo.cache-tempest-full-py3-memcached-pool https://zuul.opendev.org/t/openstack/build/c7d916cc9c4e4662b9a45b97b802a574 : SUCCESS in 1h 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6ae1e6037b52cbf9da46236b105e6450ae50142","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-08 16:27:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5cefa06748dfc091bb1df013d5efbe574989de10","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 15:22:37.000000000","message":"Patch Set 2: Code-Review+2\n\nThis looks sensible but it will need to be rebased","accounts_in_message":[],"_revision_number":2},{"id":"c567202bbe88d4fae11289352fa9e92f546ce932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 15:23:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"be288f8528f3d20395d916ab399a0d99d1cee1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:57:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4db0979551a34cd7a05fc7e5eb133653 : SUCCESS in 1h 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fc0a3eb19c4e0b8d221e2fb4d47ae1 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc436b2cb44944a598539c73cca787d1 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b964058c22949cba86b0ff9a8f4d23a : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05f25addfc94faf8a25512c4c2b5e8c : SUCCESS in 3m 59s\n- oslo.cache-tempest-full-py3-memcached-pool https://zuul.opendev.org/t/openstack/build/4972318a0bca4d84bd7ba9f3b4920ec5 : SUCCESS in 1h 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6adf08b14637c3f5869cdec6e78820db1f2ad6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-10 15:14:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c174e148413c9663188174c8cb0d669ffc918c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 15:14:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d563a369379621e9f0b8c190788f55fd9643bc58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:38:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ba407fe86c3429daa89ca3d6a3221f5 : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a22003e0434485b82ea61fb8005120d : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98509f778cf49b5a5f3c5d88a01a6ab : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5e435fc87674d5bbee69cdf7cfb6955 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2711a8653154af293d79075e29e82f3 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"6fa220d9a796a8e387de96263a08cf3c2acc8440","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:38:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cdef0613b6e9d2014d11770c669518a052343156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:40:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260fa70dd0124a3e8a155a46aede58c5 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a23dd783b5e87a9ce50dcc77b8181ec034f7816e","revisions":{"ce297bb65d28e72c375594e4b0980bd0c482509e":{"kind":"REWORK","_number":1,"created":"2021-02-03 14:21:03.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/04/773904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/04/773904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/1"}}},"commit":{"parents":[{"commit":"3b9831745fa1ee047813f7fc853a1259a6f41308","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/3b9831745fa1ee047813f7fc853a1259a6f41308"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-05-04 21:44:30.000000000","tz":-180},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-03 14:21:03.000000000","tz":0},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nChange-Id: Ief83bd490826d90eb5239ebb71394aa516f033b7\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n(cherry picked from commit c0e7543ca607b117004f836ca0270fd77e50c716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/ce297bb65d28e72c375594e4b0980bd0c482509e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/ce297bb65d28e72c375594e4b0980bd0c482509e"}]},"branch":"refs/heads/stable/ussuri"},"0efbc7e5599c282f1ee6e3070b3a1fb62c061649":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-05 11:25:33.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/04/773904/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/04/773904/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/2"}}},"commit":{"parents":[{"commit":"170bbd38e6c1158cfadb425efe050ca513271c0d","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/170bbd38e6c1158cfadb425efe050ca513271c0d"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-05-04 21:44:30.000000000","tz":-180},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-05 11:25:15.000000000","tz":60},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nChange-Id: Ief83bd490826d90eb5239ebb71394aa516f033b7\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n(cherry picked from commit c0e7543ca607b117004f836ca0270fd77e50c716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/0efbc7e5599c282f1ee6e3070b3a1fb62c061649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/0efbc7e5599c282f1ee6e3070b3a1fb62c061649"}]},"branch":"refs/heads/stable/ussuri"},"a23dd783b5e87a9ce50dcc77b8181ec034f7816e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-09 15:23:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/773904/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.cache","ref":"refs/changes/04/773904/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.cache refs/changes/04/773904/3"}}},"commit":{"parents":[{"commit":"ee7981b77d7e2f9e29ca50495a44b53708ece5f0","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/ee7981b77d7e2f9e29ca50495a44b53708ece5f0"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-05-04 21:44:30.000000000","tz":-180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-09 15:15:45.000000000","tz":0},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nChange-Id: Ief83bd490826d90eb5239ebb71394aa516f033b7\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n(cherry picked from commit c0e7543ca607b117004f836ca0270fd77e50c716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/a23dd783b5e87a9ce50dcc77b8181ec034f7816e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.cache/commit/a23dd783b5e87a9ce50dcc77b8181ec034f7816e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
