)]}'
{"id":"openstack%2Foslo.concurrency~891797","triplet_id":"openstack%2Foslo.concurrency~master~I6b40505ac609f2b06bb1087753c22dc3cee4d8db","project":"openstack/oslo.concurrency","branch":"master","topic":"release-health-check-cwi-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-21 19:47:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6b40505ac609f2b06bb1087753c22dc3cee4d8db","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-08-17 14:59:29.000000000","updated":"2023-08-21 19:47:49.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f2dd1b73ea467a42ae46f84afba5d7501829fb3","_number":891797,"virtual_id_number":891797,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-21 02:35:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 18:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 00:44:06.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"b56b9c72c99d3e14ef2b4837cebef15ea300c538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-17 14:59:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be2afb1e05628016296606975545c561f69afa92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 18:31:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90572e76e6674fc6bd874149c7d82b9b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ec790e30e6b43739c114cb55914abd2 : FAILURE in 1h 06m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26436e59d6474b008d13590306a1489b : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ff624033c3476886a097d28cc79bea : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02666cb01c9746618fa6fafa72b04c3b : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/084833c7b4834c57b02bcb16252f0de3 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02366a86c27645c0a473e85c06fd4187 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabd4dace1104e8083a06a0ca78e3996 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ce7230b1cbf46dfd3a0eaa216f9b17e061d6b25b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-21 00:44:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0c847ab1ec2f4f1d1b3df85bca619a9139a489e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 02:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9d6aa4905c47798f138cb96d1a9598\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9c450954c1944a9831f2f89a76c16d6 : SUCCESS in 1h 45m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e00a20edbec1444aabfd7aac0617a238 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bcafc77e9ab4a458c058a0c2bbb737d : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41a42f6037334336b7e92b6ef75d3347 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31e35ca7a0154873b3b0317c28c2b32b : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01afa9c628574a53a8f1a3fd3a069bd4 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cff37a047c84f729462f0c7df125ba4 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242065f9882248ed976bf8bef8e7b5ef : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0f2dd1b73ea467a42ae46f84afba5d7501829fb3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-21 19:47:49.000000000","message":"Abandoned\n\nIt seems that there was a temporary gate issue which is now disappeared, gate looks healthy, abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84cc07ff72bbe1470254ebe16c640ec2f6a50263","revisions":{"84cc07ff72bbe1470254ebe16c640ec2f6a50263":{"kind":"REWORK","_number":1,"created":"2023-08-17 14:59:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/97/891797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.concurrency","ref":"refs/changes/97/891797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.concurrency refs/changes/97/891797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.concurrency refs/changes/97/891797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.concurrency refs/changes/97/891797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.concurrency refs/changes/97/891797/1"}}},"commit":{"parents":[{"commit":"1abc8e0c5005175d0cfcd5c7ae3f6852ae148ef4","subject":"Merge \"Bump bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.concurrency/commit/1abc8e0c5005175d0cfcd5c7ae3f6852ae148ef4"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-17 14:59:16.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-17 14:59:27.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I6b40505ac609f2b06bb1087753c22dc3cee4d8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.concurrency/commit/84cc07ff72bbe1470254ebe16c640ec2f6a50263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.concurrency/commit/84cc07ff72bbe1470254ebe16c640ec2f6a50263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
