)]}'
{"id":"openstack%2Foslo.config~649753","triplet_id":"openstack%2Foslo.config~master~I7afc9ce979cebe5427a19cbf33c412ab9b23d3ae","project":"openstack/oslo.config","branch":"master","topic":"opt-help-clarification","hashtags":[],"change_id":"I7afc9ce979cebe5427a19cbf33c412ab9b23d3ae","subject":"Clarify help on config-file and config-dir options","status":"MERGED","created":"2019-04-03 22:29:17.000000000","updated":"2019-04-22 15:01:33.000000000","submitted":"2019-04-22 15:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649753-1555945293925-577aa814","meta_rev_id":"1a06ae7948b4511f8be73123965aa9487b54ef0e","_number":649753,"virtual_id_number":649753,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-16 18:45:13.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-04-22 15:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-04-17 08:49:19.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2019-04-22 13:13:34.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-04-04 14:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2019-04-22 13:13:34.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-04 14:47:52.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-04-17 08:49:19.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2019-04-22 13:13:34.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-04-22 15:01:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97881dc3d8dbb853fc18338fc3dbc3fa8cf00ae7","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-03 22:29:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ef8b80e508ccd73ac8830b750eb49da018bf058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 00:23:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/53/649753/1/check/tempest-full/670ef09/ : SUCCESS in 1h 44m 31s\n- tempest-full-py3 http://logs.openstack.org/53/649753/1/check/tempest-full-py3/fdf3eee/ : SUCCESS in 1h 32m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/649753/1/check/openstack-tox-lower-constraints/dc2d25f/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/53/649753/1/check/openstack-tox-pep8/f3200c7/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/53/649753/1/check/openstack-tox-py27/4077110/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/53/649753/1/check/openstack-tox-py35/e019a0e/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/53/649753/1/check/openstack-tox-py36/d491b63/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/53/649753/1/check/openstack-tox-py37/963abbf/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/53/649753/1/check/openstack-tox-docs/0ead085/html/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e8734922e99c42c5cd5f55cf7354b54270ac1b00","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-04-04 14:47:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23ededf00de6685596edc7df124e0bfc5c0f7370","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-04-08 09:40:12.000000000","message":"Patch Set 1: Code-Review-1\n\nShouldn\u0027t we provide some Opt param that prevents from loading values from specific sources?","accounts_in_message":[],"_revision_number":1},{"id":"518ca8512f66ab52b16bc020860b709ddbccf93f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-16 18:45:13.000000000","message":"Patch Set 1:\n\n\u003e Shouldn\u0027t we provide some Opt param that prevents from loading\n \u003e values from specific sources?\n\nI\u0027m not sure that would have general applicability. In most cases we should allow opts to be set however the user wants, but these are unique because they\u0027re used internally by oslo.config. I suppose we could add something to log a warning if they\u0027re found set in a file, but that wouldn\u0027t eliminate the need to document the behavior. I would consider that a potential followup to this change, not a replacement.","accounts_in_message":[],"_revision_number":1},{"id":"7011274c404534b2d07c9b8aeec3eba578373bc5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-04-17 08:49:19.000000000","message":"Patch Set 1: Code-Review+2\n\nSounds good to me.","accounts_in_message":[],"_revision_number":1},{"id":"8bed5002dbf55ffdbf09cacfe3c3365401846448","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-22 13:13:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a60800ea82df208565664c98d9e42de72893168","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 13:13:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4439796df2c86303290297d8b99d37baa7da9fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d1f26aa7a6768e9aa47b125bcf0eee791f49d94b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:01:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/53/649753/1/gate/tempest-full/0ed31d0/ : SUCCESS in 1h 43m 58s\n- tempest-full-py3 http://logs.openstack.org/53/649753/1/gate/tempest-full-py3/dce072a/ : SUCCESS in 1h 41m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/649753/1/gate/openstack-tox-lower-constraints/fd6636d/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/53/649753/1/gate/openstack-tox-pep8/47db783/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/53/649753/1/gate/openstack-tox-py27/3e23e55/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/53/649753/1/gate/openstack-tox-py36/30a0630/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/53/649753/1/gate/openstack-tox-py37/1b19b0f/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/53/649753/1/gate/openstack-tox-docs/ac8a41b/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce7228c717943a7cf8d2daaaf14dec84ea1accd","revisions":{"fce7228c717943a7cf8d2daaaf14dec84ea1accd":{"kind":"REWORK","_number":1,"created":"2019-04-03 22:29:17.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/53/649753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/53/649753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/53/649753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/53/649753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/53/649753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/53/649753/1"}}},"commit":{"parents":[{"commit":"6c951ed373f443727a8ea2d10d346ad769285aa2","subject":"Handle collections.abc deprecations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/6c951ed373f443727a8ea2d10d346ad769285aa2"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-03 22:23:48.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-03 22:23:48.000000000","tz":0},"subject":"Clarify help on config-file and config-dir options","message":"Clarify help on config-file and config-dir options\n\nThese options are intended to be set from the command-line, as they\nare used to find the config files we will parse. There\u0027s a\nchicken-and-egg problem with setting them from those config files,\nbut since these appear in service sample configs it may not be\nimmediately obvious to users.\n\nChange-Id: I7afc9ce979cebe5427a19cbf33c412ab9b23d3ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/fce7228c717943a7cf8d2daaaf14dec84ea1accd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/fce7228c717943a7cf8d2daaaf14dec84ea1accd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
