)]}'
{"id":"openstack%2Foslo.config~654729","triplet_id":"openstack%2Foslo.config~master~I292921cbee1f24857f0ee32cc010a8e3e6f1b8ad","project":"openstack/oslo.config","branch":"master","hashtags":[],"change_id":"I292921cbee1f24857f0ee32cc010a8e3e6f1b8ad","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 05:25:34.000000000","updated":"2019-05-03 20:36:15.000000000","submitted":"2019-05-03 20:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654729-1556913679356-7175b152","meta_rev_id":"7d7c123ab0abbc06ece4a504dc8ac461f06a54a3","_number":654729,"virtual_id_number":654729,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-03 20:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 18:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-05-03 18:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 18:13:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 20:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80d7a025afe0199f32ad566bf084551d17ab1628","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:25:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c16794b5298944bed1517e16713c70b7643b520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/654729/1/check/test-release-openstack/b04a570/ : SUCCESS in 1m 50s\n- tempest-full http://logs.openstack.org/29/654729/1/check/tempest-full/5701571/ : SUCCESS in 1h 43m 02s\n- tempest-full-py3 http://logs.openstack.org/29/654729/1/check/tempest-full-py3/e87dc30/ : SUCCESS in 1h 33m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/654729/1/check/openstack-tox-lower-constraints/f747eec/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/29/654729/1/check/openstack-tox-pep8/726a0ef/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/29/654729/1/check/openstack-tox-py27/4770402/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/29/654729/1/check/openstack-tox-py36/4c9bef9/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/29/654729/1/check/openstack-tox-py37/efffbc7/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/29/654729/1/check/openstack-tox-docs/ffdb116/html/ : SUCCESS in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/29/654729/1/check/build-openstack-releasenotes/89013bd/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"942f9f666a94a14af4165cdd8adf7d62dbc522ee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 18:13:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56efe8de09ea9e97e82c25753e958e42900a5948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67b3b01928a22a51b5d9605cab9bed4a4ae497ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:01:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/654729/1/gate/test-release-openstack/5fa066e/ : SUCCESS in 1m 56s\n- tempest-full http://logs.openstack.org/29/654729/1/gate/tempest-full/90edff3/ : SUCCESS in 1h 46m 41s\n- tempest-full-py3 http://logs.openstack.org/29/654729/1/gate/tempest-full-py3/8054c18/ : SUCCESS in 1h 29m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/654729/1/gate/openstack-tox-lower-constraints/32efd10/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/29/654729/1/gate/openstack-tox-pep8/3b7f8a0/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/29/654729/1/gate/openstack-tox-py27/3bd031c/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/29/654729/1/gate/openstack-tox-py36/77e65dc/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/29/654729/1/gate/openstack-tox-py37/f5d1d1f/ : SUCCESS in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/29/654729/1/gate/openstack-tox-docs/1d49014/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/29/654729/1/gate/build-openstack-releasenotes/2f86264/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"80b6af7f9f33fbf16b682d73850b35c32118c4f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:01:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ebf33938803767d74c154bdb5dec4ab3ab7415de","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 20:36:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d7c123ab0abbc06ece4a504dc8ac461f06a54a3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 20:36:15.000000000","message":"Created a revert of this change as Ief33299e46242d9435a382116cf51c2780cc1beb","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"034a90d005549e98e5fcd6c481ec79fe1d99bc22","revisions":{"034a90d005549e98e5fcd6c481ec79fe1d99bc22":{"kind":"REWORK","_number":1,"created":"2019-04-23 05:25:34.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/29/654729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/29/654729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/29/654729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/29/654729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/29/654729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/29/654729/1"}}},"commit":{"parents":[{"commit":"0bc22839b09fcd7ae2de4f7fe5d63ac055fd0a78","subject":"Merge \"Clarify help on config-file and config-dir options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/0bc22839b09fcd7ae2de4f7fe5d63ac055fd0a78"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:30.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I292921cbee1f24857f0ee32cc010a8e3e6f1b8ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/034a90d005549e98e5fcd6c481ec79fe1d99bc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/034a90d005549e98e5fcd6c481ec79fe1d99bc22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
