)]}'
{"id":"openstack%2Foslo.config~669459","triplet_id":"openstack%2Foslo.config~master~Ia547c6f2e27c12ef58eb955286b5f2ab5359c2e2","project":"openstack/oslo.config","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ia547c6f2e27c12ef58eb955286b5f2ab5359c2e2","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:43:07.000000000","updated":"2019-07-11 13:16:05.000000000","submitted":"2019-07-11 13:16:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669459-1562850965641-77280bbb","meta_rev_id":"0d23d9499d251a4a1c4112aa1aa93997f4a29201","_number":669459,"virtual_id_number":669459,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-11 13:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-07-08 14:35:14.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-11 11:21:53.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-07-11 11:21:53.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 14:35:14.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-11 11:21:53.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-11 13:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e8e15dd7b5d550883d2b59513dde48779f31dee","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:43:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c7c42fed6cd19576adbfb0743de2d4e976421eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/59/669459/1/check/test-release-openstack/1278e1c/ : SUCCESS in 2m 27s\n- tempest-full http://logs.openstack.org/59/669459/1/check/tempest-full/2fc5434/ : SUCCESS in 1h 58m 58s\n- tempest-full-py3 http://logs.openstack.org/59/669459/1/check/tempest-full-py3/f44664e/ : SUCCESS in 1h 39m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/669459/1/check/openstack-tox-lower-constraints/727dc33/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/59/669459/1/check/openstack-tox-pep8/46dd44c/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/59/669459/1/check/openstack-tox-py27/8daa4e1/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/59/669459/1/check/openstack-tox-py36/15fdeac/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/59/669459/1/check/openstack-tox-py37/8e0dca2/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/59/669459/1/check/openstack-tox-docs/65b1193/html/ : SUCCESS in 3m 25s\n- build-openstack-releasenotes http://logs.openstack.org/59/669459/1/check/build-openstack-releasenotes/269dcb6/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"32a4a28a6da47d516c31dfc4b24b8ca6086a8e1d","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-08 14:35:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28493090ffdd7c3afb47734b1862bcaa3f614ee0","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-11 11:21:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a80bc21b02482862a8bd646ff19ec9ca278625cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:22:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"293f9a60cb43e0c3a0e1f313ef348d64af90a61d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:16:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b977da19d2d1486697ba08d21d06bfcc862b842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:16:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/59/669459/1/gate/test-release-openstack/516ca88/ : SUCCESS in 3m 00s\n- tempest-full http://logs.openstack.org/59/669459/1/gate/tempest-full/cad5957/ : SUCCESS in 1h 47m 52s\n- tempest-full-py3 http://logs.openstack.org/59/669459/1/gate/tempest-full-py3/d3374d6/ : SUCCESS in 1h 33m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/669459/1/gate/openstack-tox-lower-constraints/4f66355/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/59/669459/1/gate/openstack-tox-pep8/31cc7a5/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/59/669459/1/gate/openstack-tox-py27/5266bab/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/59/669459/1/gate/openstack-tox-py36/482af17/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/59/669459/1/gate/openstack-tox-py37/c5f6b5c/ : SUCCESS in 10m 33s\n- openstack-tox-docs http://logs.openstack.org/59/669459/1/gate/openstack-tox-docs/ff61922/html/ : SUCCESS in 6m 02s\n- build-openstack-releasenotes http://logs.openstack.org/59/669459/1/gate/build-openstack-releasenotes/d529636/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2488c1e1cee151a3a65c60f371f0bea2f15d4bca","revisions":{"2488c1e1cee151a3a65c60f371f0bea2f15d4bca":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:43:07.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/669459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/59/669459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/59/669459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/59/669459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/59/669459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/59/669459/1"}}},"commit":{"parents":[{"commit":"dea9af2fe91e3abe6ca88ea8739cff023191abb9","subject":"Merge \"Add special handling for keystone_authtoken group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/dea9af2fe91e3abe6ca88ea8739cff023191abb9"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:47.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:43:07.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ia547c6f2e27c12ef58eb955286b5f2ab5359c2e2\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/2488c1e1cee151a3a65c60f371f0bea2f15d4bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/2488c1e1cee151a3a65c60f371f0bea2f15d4bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
