)]}'
{"id":"openstack%2Foslo.config~694301","triplet_id":"openstack%2Foslo.config~master~Ibcdd3a9f7e7ea421e906cf1d96f95b516008b5d7","project":"openstack/oslo.config","branch":"master","topic":"stop_configuring_install_command","hashtags":[],"change_id":"Ibcdd3a9f7e7ea421e906cf1d96f95b516008b5d7","subject":"Stop configuring install_command in tox.","status":"MERGED","created":"2019-11-14 14:06:11.000000000","updated":"2019-11-14 17:53:36.000000000","submitted":"2019-11-14 17:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694301-1573753879814-ee4c743d","meta_rev_id":"20daa68930239861688464ace0a1159e80ebb480","_number":694301,"virtual_id_number":694301,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 17:51:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-11-14 14:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-11-14 14:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-11-14 14:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 14:07:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-14 14:44:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-11-14 17:51:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00d3c7cbf3dec4b6ccb81666563e68cff17c9c7e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-14 14:06:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c731fa3af51a2bd169fff92da33e19238c509ac2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-14 14:07:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8542103eec31d02cd21e33c25d550659cd02ba18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-14 14:44:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n@hberaud Feel free to just fast approve these if there are any more :)","accounts_in_message":[],"_revision_number":1},{"id":"7cd2b959f1ed5491d4cc6ad88cdc8abea7747298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:03:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c523366754c743d996bdd618035289f2 : SUCCESS in 1h 56m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/048e87b138b8438d81320bae1ae5d5a6 : SUCCESS in 1h 33m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64486aba08a9457ebca0e5ef177956ce : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19d671d4c1148f6a394fe7b29031c24 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6791068672a742e4b8b7f6caf8f1da17 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d56974f973d4388b3df8e17bf1184dc : SUCCESS in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbee8f3952ad4637bcfee60574a51ee8 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b95fe2063314832b3ec60efed8d6cff : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c720ccccb0a4c0c8bd9b26baac22e7a : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c33f8395c6600c09ba806d8d5fa0f6c9f78be44b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:03:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e428d98f8afc38bdbcb02bb9e77bb5bd594eaec2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:51:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a3147c31328a28adccde00d3754c45c66460c01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:51:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1df03108398b4019aee891c9cbac0ca3 : SUCCESS in 1h 42m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/283411a685cd42db9078c6f52ad9440b : SUCCESS in 1h 24m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55dae9ae13bb465eb0b04a6dea940906 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e15322eff894451a8f34acdd24a4a9a : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f10396e9d4924b8e9b2d77a538b97509 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38566fea585449690109c6a20823915 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d59ede21012847f187f30ffc1b9dac83 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad58a63667b24abb97274f27bf348a22 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7461fcf33b79402ca282f9dbc31e365b : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a5637c8f119945fa8bfed7d9e62b7951e038fed5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db2234153954605af3ad457a3102ccf : SUCCESS in 1m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f67441d34d940438b4a0cdfdca97031 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0c992f5c088054196389ab12599e52ec5970d3a","revisions":{"f0c992f5c088054196389ab12599e52ec5970d3a":{"kind":"REWORK","_number":1,"created":"2019-11-14 14:06:11.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/01/694301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/01/694301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/01/694301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/01/694301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/01/694301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/01/694301/1"}}},"commit":{"parents":[{"commit":"561db7ac328417cb9091a21798b8b9e9def35e6a","subject":"fix: fix float opt min and max value format errror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/561db7ac328417cb9091a21798b8b9e9def35e6a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-14 14:05:29.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-14 14:05:29.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: Ibcdd3a9f7e7ea421e906cf1d96f95b516008b5d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/f0c992f5c088054196389ab12599e52ec5970d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/f0c992f5c088054196389ab12599e52ec5970d3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
