)]}'
{"id":"openstack%2Foslo.config~699522","triplet_id":"openstack%2Foslo.config~stable%2Ftrain~Id3b28465d645a24f0fcebff2dd68a9bd30e21594","project":"openstack/oslo.config","branch":"stable/train","topic":"bug/1856312","hashtags":[],"change_id":"Id3b28465d645a24f0fcebff2dd68a9bd30e21594","subject":"Ensure option groups don\u0027t change during logging","status":"MERGED","created":"2019-12-18 00:52:35.000000000","updated":"2019-12-18 16:54:07.000000000","submitted":"2019-12-18 16:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699522-1576687943831-cc347a4a","meta_rev_id":"2177d76eef44fc1047d5c349396fa876117a9415","_number":699522,"virtual_id_number":699522,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 16:52:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-12-18 15:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-18 02:14:12.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-18 15:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 02:14:12.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-18 15:07:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-18 16:52:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51fa528b1b27e827d692acbab22a5071bbc120b7","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2019-12-18 00:52:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aacdcb946d50de655abdd93b8241cde6a5613ac5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2019-12-18 00:52:42.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 5b83858ea8e36916eb7ca046ccc59a44c9ab7eb1","accounts_in_message":[],"_revision_number":1},{"id":"d029419a708823442dcb77b5b2d9f6e93ada5d60","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-18 02:13:37.000000000","message":"Topic changed from bug/1856312-stable/train to bug/1856312","accounts_in_message":[],"_revision_number":1},{"id":"552a4e0b0e3307a6d0f5e9a71258be5f4ea7bca6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-18 02:14:12.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport, the change is merged on master, lgtm.","accounts_in_message":[],"_revision_number":1},{"id":"5c5e01db2bfefcf89ff21c74c64e4dd47c9aaf10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:33:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0cba05774c1446eda77489fdc24812c0 : SUCCESS in 1h 30m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77eb388515d6437a8a893fd1e030015a : SUCCESS in 1h 39m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab91bc6879944fd191c40559cef58a2e : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5197a6dde5a4c2bb52be5ec4eec7c6b : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/725971e37ac1486cbc09402a08fd39ad : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd204dc7fe3343ae8878e3176b78cb1a : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b5686587c01436395c63f22916fc25d : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057639deeaf44d1fa1d740c8ca961d5c : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"620d9b62a220f7f0b11a58eb944328f92297f016","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 15:07:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9158c420fd7603be858d8e1f02704437de32b559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 15:07:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f9b81301dde60c5b641ee7092f6d104f358cd0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:52:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b01b9e271910286958cf4b425cb28ce6b59b049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:52:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/db174317c53d43aa812cf92c6ac45ef0 : SUCCESS in 1h 38m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/672a8fcc66a74cb7b1af29f4fe83486c : SUCCESS in 1h 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9ad70eb94b743d4bf616b463a868c0d : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81af4e5216ec4d55b9f7afbb4ced11a7 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/854bf1bf65ff484482e5f10e39f67e13 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c78b879c5f4cf5821398ae6618040f : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37c6fa86d459433abe627ff67ee34035 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e81e76e06a8421bb924ba208447a15d : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2177d76eef44fc1047d5c349396fa876117a9415","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcede9450814c178f399f90a97c4917 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22c286c63b15b6f0acf6f696b98226a9257bf745","revisions":{"22c286c63b15b6f0acf6f696b98226a9257bf745":{"kind":"REWORK","_number":1,"created":"2019-12-18 00:52:35.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/22/699522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/22/699522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/22/699522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/22/699522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/22/699522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/22/699522/1"}}},"commit":{"parents":[{"commit":"d844a0ea8f03e8290363baa74bee7c14bac4e335","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/d844a0ea8f03e8290363baa74bee7c14bac4e335"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:18:33.000000000","tz":0},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2019-12-18 00:52:35.000000000","tz":0},"subject":"Ensure option groups don\u0027t change during logging","message":"Ensure option groups don\u0027t change during logging\n\noslo.config allows us to configure groups and options dynamically. This\ncan cause a race with our logging as we attempt to iterate through\noption groups that are changing under our feet. This wouldn\u0027t be a huge\nissue, since these are just logs are we can always log again, if needed,\nbut we store groups in a dictionary and Python doesn\u0027t like us changing\nthe size of a dict it\u0027s iterating through:\n\n  RuntimeError: dictionary changed size during iteration\n\nGiven that we\u0027re only reading through this option group and don\u0027t need\nto worry about a group _disappearing_, the solution is pretty simple:\ncreate a copy of our option group names ahead of time so we don\u0027t need\nto worry about new ones coming and messing things up.\n\nNo tests are included since this is a race and the only way I see to\nreproduce this would involve lots of ugly threading.\n\nChange-Id: Id3b28465d645a24f0fcebff2dd68a9bd30e21594\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1856312\n(cherry picked from commit e3e2ba55eeeb86a9bc0624bb2592e46583e839e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/22c286c63b15b6f0acf6f696b98226a9257bf745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/22c286c63b15b6f0acf6f696b98226a9257bf745"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
