)]}'
{"id":"openstack%2Foslo.config~727236","triplet_id":"openstack%2Foslo.config~master~I2a539263e8cfc7d15243d78e20b70b8ed763b8fd","project":"openstack/oslo.config","branch":"master","topic":"flake8","hashtags":[],"change_id":"I2a539263e8cfc7d15243d78e20b70b8ed763b8fd","subject":"Cap flake8 to be \u003c 3.8.0","status":"MERGED","created":"2020-05-12 15:39:37.000000000","updated":"2020-05-14 21:17:26.000000000","submitted":"2020-05-14 21:15:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727236-1589490944909-78abda74","meta_rev_id":"70cc3a70524d9d590e687a3219f97012ee0d65e8","_number":727236,"virtual_id_number":727236,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 21:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-14 19:41:47.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-05-12 16:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-14 19:41:47.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 16:52:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-14 19:41:47.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-14 21:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ffc484648b7ab9d862cc616e7c1c63c1cf79478","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 15:39:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7595cb263ff5fcf362b0dd7b370d4b41affe0980","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 15:40:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bc7ba0a82f602aad4530f5e680ed19d189b02b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 16:21:12.000000000","message":"Patch Set 2: Workflow-1\n\nWell that didn\u0027t work right.","accounts_in_message":[],"_revision_number":2},{"id":"1462ccd345c4a623d15203e4b938d4caf2ba94e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 16:22:17.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b4070e0aa63d2b63a2263fddd4f9ff4552ea9aa1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-12 16:52:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fbdcec58c714fcdfe25b5db9d3077cd70add2e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:00:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e1f02deb1814a488dae90815260736f : SUCCESS in 5m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91351c7a20174846877098f0e547ad56 : SUCCESS in 1h 33m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cce1e3bcdde744c5a4ecdf198e09adc5 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a934a4cac444582ae0b66eb0f91db2e : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c15ce91e104d57874acce4d2646f44 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fac63f0baf3b4f59a9aa3df00f997192 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7375f23877324b6ea853a09b258d8f14 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c9771ccdd7a46a7afd6d380077cee36 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"97ab1729fa2923442898a480301a7a9794fb66fe","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-14 19:41:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nlgtm as a backportable workaround for this.","accounts_in_message":[],"_revision_number":3},{"id":"7aee2a780a5c333bc47a00f376fe06a166192295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:41:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"772561f2a38742976469894fec9510d9ea15962e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:15:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/32f4872d80be4e8ca42c0920c9a65b16 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/299e53bd164045eda1bbfacff97cc3c0 : SUCCESS in 1h 33m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96e864d63a54611a345820667b55cda : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c88ff79508d4c31809bd9b185ea0397 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3523813e1448ea980b7118530ac621 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618eda3034384ac494290cc2568e967f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3a96a5ad594cce81a25c25327e33b0 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a70048117b484ed5ba06f05127d91a25 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"e818ef7a5b7298fd70967a2d63ed6793390f45cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:15:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"75885d5e3cdf16cc7df856c14c59d93c21ce7f3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:17:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ac47fc2ed74fe0b0a7d2c8f8e96c5c : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806aa617334642498cbdc7d1b7f999d2 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1836ad4d2f091fa1825af91c09b3064012997dd","revisions":{"7c427bc6bd3cef3cf6aecd317c0b582d167c9fe9":{"kind":"REWORK","_number":1,"created":"2020-05-12 15:39:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/727236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/36/727236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/1"}}},"commit":{"parents":[{"commit":"2727d3e4e1f08a7d06d6b782ee9ffee85248ac49","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/2727d3e4e1f08a7d06d6b782ee9ffee85248ac49"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 15:37:38.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 15:37:38.000000000","tz":-300},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures with the\ncurrent code. To avoid that, this caps flake8 to a version that had been\ntested withouth errors so hacking does not pull in a newer version.\n\nWe will likely want to raise hacking to 3.x and drop this cap, but to\nmake it easy to backport to stable branches, taking this approach first.\n\nChange-Id: I2a539263e8cfc7d15243d78e20b70b8ed763b8fd\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/7c427bc6bd3cef3cf6aecd317c0b582d167c9fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/7c427bc6bd3cef3cf6aecd317c0b582d167c9fe9"}]},"branch":"refs/heads/master"},"1dd5913538fbd23086e71670465202a72050b50b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-05-12 15:40:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/727236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/36/727236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/2"}}},"commit":{"parents":[{"commit":"d93bfe182e46837e0f90a3bb2e73aa3aeec8fdc9","subject":"Merge \"Use stevedore to list the source drivers in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/d93bfe182e46837e0f90a3bb2e73aa3aeec8fdc9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 15:37:38.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 15:40:02.000000000","tz":-300},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures with the\ncurrent code. To avoid that, this caps flake8 to a version that had been\ntested withouth errors so hacking does not pull in a newer version.\n\nWe will likely want to raise hacking to 3.x and drop this cap, but to\nmake it easy to backport to stable branches, taking this approach first.\n\nChange-Id: I2a539263e8cfc7d15243d78e20b70b8ed763b8fd\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 7c427bc6bd3cef3cf6aecd317c0b582d167c9fe9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1dd5913538fbd23086e71670465202a72050b50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1dd5913538fbd23086e71670465202a72050b50b"}]},"branch":"refs/heads/master"},"b1836ad4d2f091fa1825af91c09b3064012997dd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-12 16:22:17.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/727236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/36/727236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/36/727236/3"}}},"commit":{"parents":[{"commit":"d93bfe182e46837e0f90a3bb2e73aa3aeec8fdc9","subject":"Merge \"Use stevedore to list the source drivers in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/d93bfe182e46837e0f90a3bb2e73aa3aeec8fdc9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 15:37:38.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 16:22:03.000000000","tz":-300},"subject":"Cap flake8 to be \u003c 3.8.0","message":"Cap flake8 to be \u003c 3.8.0\n\nThe 3.8.0 version introduces some checks that cause failures with the\ncurrent code. To avoid that, this caps flake8 to a version that had been\ntested withouth errors so hacking does not pull in a newer version.\n\nWe will likely want to raise hacking to 3.x and drop this cap, but to\nmake it easy to backport to stable branches, taking this approach first.\n\nChange-Id: I2a539263e8cfc7d15243d78e20b70b8ed763b8fd\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/b1836ad4d2f091fa1825af91c09b3064012997dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/b1836ad4d2f091fa1825af91c09b3064012997dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
