)]}'
{"id":"openstack%2Foslo.config~744307","triplet_id":"openstack%2Foslo.config~master~I95b3ec55f2e056aa42ef207c40e822cbb639a812","project":"openstack/oslo.config","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I95b3ec55f2e056aa42ef207c40e822cbb639a812","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 22:25:00.000000000","updated":"2020-08-02 23:30:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a61495117f93483dcec95b9352b512b8e086e487","_number":744307,"virtual_id_number":744307,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 21:17:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 21:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a88b03b612198521a5e295dd1fd4d9daf1db5183","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 22:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18895a982ba0c651e0ad43c266d29ba1d4db25a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 23:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d594fb890b64012be9d8b11aa43513a : SUCCESS in 1h 15m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4305ce2d46a4487080d548706ed0be95 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67195c8f23294c15b997cbf7bf996681 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93259bf22f6c468fa95e87b87830fc24 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/007d1fd7b7c04f809a97b9a596278520 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5205329cd9334c0e80fabf788fcb99e1 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f156c7178932fe3b527bb10613a99180e5992592","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:51:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"97a9866eacb194cd1fc21ddaa86d32537dca3bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:17:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a0b1482330741cf81c2ad44b57f7842 : SUCCESS in 1h 08m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b9c0fba84d040568f47927014922473 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad3f7cfe87642d8a8291d76b4d1b8fa : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/266c0eeec9dd41159441a848cc9d8c2b : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e906faefa7ce4d4e93cab9775fc4e847 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68595c9926a047c59f8514663e58d8cd : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"769fd70d0b41064f0a9f19c1ae013660e16b255c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:30:17.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20fad99185bd25ecbc0569a276e4ea04289184d7","revisions":{"367b2f4ea4f5979b092dde0323044c3e703d97c5":{"kind":"REWORK","_number":1,"created":"2020-08-01 22:25:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/744307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/07/744307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/1"}}},"commit":{"parents":[{"commit":"a2652b3584da06e4c4ebe96261672309b01c1122","subject":"Fix unit tests to work with stevedore \u003e 2.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/a2652b3584da06e4c4ebe96261672309b01c1122"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:24:27.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:25:01.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: I95b3ec55f2e056aa42ef207c40e822cbb639a812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/367b2f4ea4f5979b092dde0323044c3e703d97c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/367b2f4ea4f5979b092dde0323044c3e703d97c5"}]},"branch":"refs/heads/master"},"20fad99185bd25ecbc0569a276e4ea04289184d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 19:51:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/744307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/07/744307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/07/744307/2"}}},"commit":{"parents":[{"commit":"a2652b3584da06e4c4ebe96261672309b01c1122","subject":"Fix unit tests to work with stevedore \u003e 2.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/a2652b3584da06e4c4ebe96261672309b01c1122"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:24:27.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 19:51:06.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I95b3ec55f2e056aa42ef207c40e822cbb639a812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/20fad99185bd25ecbc0569a276e4ea04289184d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/20fad99185bd25ecbc0569a276e4ea04289184d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
