)]}'
{"id":"openstack%2Foslo.config~768424","triplet_id":"openstack%2Foslo.config~master~I5e1cd8f4c19390d645b382a7deb4e36702b99932","project":"openstack/oslo.config","branch":"master","hashtags":[],"change_id":"I5e1cd8f4c19390d645b382a7deb4e36702b99932","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-12-24 06:41:13.000000000","updated":"2021-04-15 14:07:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1248c8a4689a83127bfedece91f356b6a48f5da3","_number":768424,"virtual_id_number":768424,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-24 08:19:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":-1,"date":"2020-12-25 14:36:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-25 10:37:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 08:19:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-25 10:37:39.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-12-25 14:36:28.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"}],"messages":[{"id":"35fb699f34e406e80d405bd3038134da7b8544ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-24 06:41:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb2e7274d35d24f897ca3eac0c61d5f35a3880ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 08:19:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9b6bfbec7b144fdb5432dc03386c986 : SUCCESS in 1h 31m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ddc4d00e87a460799207d0a046109e6 : FAILURE in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7aeaa07f264030b7676d75bd3f6cd2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/790b04c4d60540a1b6d0ad850be15c3d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f20b7ee7421d4d9caaa56e1c833ebfcc : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bb4319d2c6449938ee3090831f2c002 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e70bbd51b6f492883ab6e5c87f97bd7 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb3f0e94596646408c3b2d90c4ed0a33 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e44d47215b10395ba54dfe348e01b694e0468d38","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-25 10:37:39.000000000","message":"Patch Set 1: Code-Review-1\n\nhttps://review.opendev.org/c/openstack/oslo.config/+/761340","accounts_in_message":[],"_revision_number":1},{"id":"fd819bdacd859fc2ca7f96cd8ab8d223208218ee","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-25 14:36:28.000000000","message":"Patch Set 1: Code-Review-1\n\ndup with https://review.opendev.org/c/openstack/oslo.config/+/761340","accounts_in_message":[],"_revision_number":1},{"id":"1248c8a4689a83127bfedece91f356b6a48f5da3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 14:07:01.000000000","message":"Abandoned\n\nThis has been done elsewhere now. Thanks for the patch nonetheless!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f12d357a85c7450c2de45ff5696e8afe1d4901e","revisions":{"2f12d357a85c7450c2de45ff5696e8afe1d4901e":{"kind":"REWORK","_number":1,"created":"2020-12-24 06:41:13.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/24/768424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/24/768424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/24/768424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/24/768424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/24/768424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/24/768424/1"}}},"commit":{"parents":[{"commit":"efa46b5f441ed377a777f33b66aa4070ae133c9f","subject":"Merge \"Adding --check-defaults to validator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/efa46b5f441ed377a777f33b66aa4070ae133c9f"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 06:40:18.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 06:41:03.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I5e1cd8f4c19390d645b382a7deb4e36702b99932\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/2f12d357a85c7450c2de45ff5696e8afe1d4901e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/2f12d357a85c7450c2de45ff5696e8afe1d4901e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
