)]}'
{"id":"openstack%2Foslo.config~794816","triplet_id":"openstack%2Foslo.config~stable%2Fussuri~I70ab87c9bed093cad883b6301b8a09753fc470d9","project":"openstack/oslo.config","branch":"stable/ussuri","topic":"automated_config_validation","hashtags":[],"change_id":"I70ab87c9bed093cad883b6301b8a09753fc470d9","subject":"config-generator yaml format doesn\u0027t work with i18n fields","status":"MERGED","created":"2021-06-04 17:27:16.000000000","updated":"2021-06-09 17:28:04.000000000","submitted":"2021-06-09 17:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794816-automated_config_validation","cherry_pick_of_change":794815,"cherry_pick_of_patch_set":1,"meta_rev_id":"e55434438ec52e895d3246a8d2b25e04adf1aa3e","_number":794816,"virtual_id_number":794816,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 17:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-09 14:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-09 16:02:19.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2021-06-09 12:32:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-06-09 16:02:19.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2021-06-06 14:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-04 17:27:16.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43c836a4ae327778c187ca30c44265febd2dcaf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-04 17:27:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"1dca8e036c339b30800b2c567516c1767754aefa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 18:43:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a17e030377894fe7a19e77fff01d4f4c : SUCCESS in 1h 13m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5de314489511413fbda8a12639432943 : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f4a1ebc07940c99448c46d9460b6f5 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf895670a05041dc94d69934cc4b0f0f : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/970da3b2afde40bb8a01c8fbf4e72d47 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25a903f916c4f3db9adf745202b3186 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d4358e4354f9ef5d5d230bcc97c81653520cf588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-06-05 14:12:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a0d4905ab8cc7a64419b89fe872c0a3948b67161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 15:56:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d352499838743a6ab0464c0820f4623 : SUCCESS in 1h 35m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89ce15d1b7b740058b01da7a87ac9f77 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ed02dab5ed40ef997db67a831c49ef : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff92f86ea7db41c9afbae30adc13f6d6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da497c1b666e468995a74589c0e60164 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f012883de624fd0b5f38efa51d59cf5 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c11e6b172064c9b393d6e32b8ea7197aa808c432","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-06 14:31:40.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"6d53ad291de11a9332670c349f3169f29074fac3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 12:32:53.000000000","message":"Patch Set 2: Code-Review+1\n\nClean backport. Need Victoria merged first (only +1)","accounts_in_message":[],"_revision_number":2},{"id":"46922672c12c0f5d8e7a57b81b5dfa04154bd2ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-09 14:49:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c07dc6d8c9f0c51281350856ddd096d776898ef0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-06-09 16:02:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nVictoria is merged.","accounts_in_message":[],"_revision_number":2},{"id":"e370ec819662adc4d28ab7ec04ec3715a6bed62d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:02:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"accae8c74953a943eded41e3ee4b804c8e643d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 17:26:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d180299ed3e47a2b8153357f521ca3d : SUCCESS in 1h 22m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08dcdebdac5d4a3a97f7ca5cb1b40976 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f4d9fe0f144bd79617e63d785a8e1a : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61bc9e6b51c4f88937b37eca99472de : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5b6bdc397b74a0a89298879bbefc98c : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d779d6f4830414d833005d39303c325 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4b74bbe73947fd9ff8ec3599dd245f7204754535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 17:26:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e55434438ec52e895d3246a8d2b25e04adf1aa3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 17:28:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d621c7d22874d9c8a24d821b3921e99 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be3558bb51143c268c0e607e6189b496efbbf308","revisions":{"11f8a430d34ed880a02eaa59092e8c9873f37c6c":{"kind":"REWORK","_number":1,"created":"2021-06-04 17:27:16.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/16/794816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/16/794816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/1"}}},"commit":{"parents":[{"commit":"1fcb3e8acd3c229c74c87d80fd183cfc2432bace","subject":"Add a new type HostDomain.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1fcb3e8acd3c229c74c87d80fd183cfc2432bace"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-12 01:49:04.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-04 17:27:16.000000000","tz":0},"subject":"config-generator yaml format doesn\u0027t work with i18n fields","message":"config-generator yaml format doesn\u0027t work with i18n fields\n\nThis is because there\u0027s no yaml representer for i18n Messages object.\nThis patch aims to add this representer and allow the generation of\nconfigurations using oslo.i18n strings.\n\nOne example of this is cinder.\n\nCloses-bug: #1928582\nChange-Id: I70ab87c9bed093cad883b6301b8a09753fc470d9\n(cherry picked from commit e5fc313ecf47813364a2111b987ced823c0d64e3)\n(cherry picked from commit de1dbeed235d2df7cf1528aef6ad4a70ddf8559f)\n(cherry picked from commit 0950f82ec59e1b4a2af66b5c5940a651a5cb574f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/11f8a430d34ed880a02eaa59092e8c9873f37c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/11f8a430d34ed880a02eaa59092e8c9873f37c6c"}]},"branch":"refs/heads/stable/ussuri"},"be3558bb51143c268c0e607e6189b496efbbf308":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-05 14:12:38.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/16/794816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/16/794816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/16/794816/2"}}},"commit":{"parents":[{"commit":"f1ab25875d8540f69d54bd6dfe1e1b74da823fd4","subject":"Updating lower-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/f1ab25875d8540f69d54bd6dfe1e1b74da823fd4"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-05-12 01:49:04.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-06-05 14:12:38.000000000","tz":0},"subject":"config-generator yaml format doesn\u0027t work with i18n fields","message":"config-generator yaml format doesn\u0027t work with i18n fields\n\nThis is because there\u0027s no yaml representer for i18n Messages object.\nThis patch aims to add this representer and allow the generation of\nconfigurations using oslo.i18n strings.\n\nOne example of this is cinder.\n\nCloses-bug: #1928582\nChange-Id: I70ab87c9bed093cad883b6301b8a09753fc470d9\n(cherry picked from commit e5fc313ecf47813364a2111b987ced823c0d64e3)\n(cherry picked from commit de1dbeed235d2df7cf1528aef6ad4a70ddf8559f)\n(cherry picked from commit 0950f82ec59e1b4a2af66b5c5940a651a5cb574f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/be3558bb51143c268c0e607e6189b496efbbf308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/be3558bb51143c268c0e607e6189b496efbbf308"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
