)]}'
{"id":"openstack%2Foslo.config~823710","triplet_id":"openstack%2Foslo.config~master~I1e4ff0ba57323232736b94e1e49f9e899fc78487","project":"openstack/oslo.config","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-06-20 11:15:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1e4ff0ba57323232736b94e1e49f9e899fc78487","subject":"WIP: Add a new max_sample_default","status":"ABANDONED","created":"2022-01-06 15:49:26.000000000","updated":"2023-06-20 11:15:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"419fd5c3b8f3c0ae01b0cd9351b1f7ac401f47ad","_number":823710,"virtual_id_number":823710,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-13 12:20:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 17:38:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb2219da2b1aee3c2cc1001d6edf0395bb75ed05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-06 15:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4f10ff19b4954e79df535b5cca2c025822149a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-06 15:56:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"720c29898d2249cab4c6e44711f1571154a80594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:38:32.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/198b37d18857484da6704d845eac9540 : SUCCESS in 1h 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b6d6228f6a4e6b999ef32678941b0c : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27edb8707bc1432a8b8649e3b45bac28 : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b2a46d0115d4088afef05297010530d : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f623d29d32cd416ea2eb91835a85e733 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/880a698f607d499584dab265326d3c81 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7dfcb7cb27050cfb664292d665b1dcbd571ebe3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-10 14:47:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1003269d0a47deccd967d53fceae0c4900a6faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 16:23:58.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50cc83515a62434bb4cd05d14209c802 : SUCCESS in 1h 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b084cc296a4b7f87e17e52993d99c8 : FAILURE in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3eea1a6f93e4dd18a6c52cf76a57f0a : FAILURE in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8168b25bb10445fb5a674d9f4adc8b3 : FAILURE in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2f3596205c4730b39d99094b3b728c : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc3f166b0a94962890f2726d5d211ac : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"7c8be800ae53bb833b74ff266d07e4c11e7b35a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-12 12:29:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9197f9eae3a6ca5a6e8592a6cb66c5958ede0ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-12 12:33:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8748b9ddd24681408802f88034bfba645251cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 14:21:06.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42132fc473f24ba78ed3a2ebaac8bd6e : SUCCESS in 1h 36m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac67aec5fc347db8e3c4481921df2ea : FAILURE in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f138513b5c8d4c508d16ccbb79bc0ab1 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e81973619b9c4a02a0afac40150363c5 : FAILURE in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bd82b3083e4be8bd9e4d218f518a34 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/836c25d821a0413893d33755cff13aae : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":5},{"id":"88436932d364a43f87fdbd7584a15b68a26e42c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-12 22:40:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"013716f25afdffd45d3eb061de11d379166430b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-12 22:42:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0022ca5b86f0d6eb4b206ac007fab40ab4590a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 00:25:28.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95bd7b739d9d4c0798c1d4c6a6a940ae : SUCCESS in 1h 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef4fc6b0c8042ef9c6779db16ae9520 : FAILURE in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c2cd46be894f9eb4be35dea2685831 : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83440e096aa4f33b4b97057aa5b38b2 : FAILURE in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7716441de0b4858b460f9f49c9a7c59 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b13b1d53d4e4845b3196059567845a5 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":7},{"id":"22b557bb55a3611d11bea30088447dece4d94de4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-13 10:52:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9b1f7fd8ab870e05174f9e2bae67d12ecde9243a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:20:30.000000000","message":"Patch Set 8: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e06b334545b474dbdfcfef78d7c8525 : SUCCESS in 1h 25m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164fb32a1a924bf4a65827a90222cefb : FAILURE in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7796d3a13d54deba29f890c75215764 : FAILURE in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdd99175d5de4f8b8248c8b4cb7c4844 : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68d12b7ee694f25bf9e39e8a92ec4bb : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf09912d26214214b17cf9106dc3cc3d : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":8},{"id":"419fd5c3b8f3c0ae01b0cd9351b1f7ac401f47ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-06-20 11:15:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"538b9de970946e8434c568d06ba32e0d808eb049","revisions":{"dd2a8f18038b25fef1241fc3e076f7e4a5a08656":{"kind":"REWORK","_number":1,"created":"2022-01-06 15:49:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/1"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:49:24.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/dd2a8f18038b25fef1241fc3e076f7e4a5a08656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/dd2a8f18038b25fef1241fc3e076f7e4a5a08656"}]},"branch":"refs/heads/master"},"66bdd9563aabf0547e8a2e100bc35964204c2673":{"kind":"REWORK","_number":2,"created":"2022-01-06 15:56:46.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/2"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:56:38.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/66bdd9563aabf0547e8a2e100bc35964204c2673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/66bdd9563aabf0547e8a2e100bc35964204c2673"}]},"branch":"refs/heads/master"},"9264b7583e91991c90b57e676bce16a84cf82f3b":{"kind":"REWORK","_number":3,"created":"2022-01-10 14:47:43.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/3"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-10 14:47:35.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/9264b7583e91991c90b57e676bce16a84cf82f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/9264b7583e91991c90b57e676bce16a84cf82f3b"}]},"branch":"refs/heads/master"},"18fb58036e768b9ae05bd5fde9a592f8a59caef7":{"kind":"REWORK","_number":4,"created":"2022-01-12 12:29:41.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/4"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-12 12:29:33.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/18fb58036e768b9ae05bd5fde9a592f8a59caef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/18fb58036e768b9ae05bd5fde9a592f8a59caef7"}]},"branch":"refs/heads/master"},"241307c3ea1f7c21ecc6edc438291707a8e538b0":{"kind":"REWORK","_number":5,"created":"2022-01-12 12:33:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/5"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-12 12:33:20.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/241307c3ea1f7c21ecc6edc438291707a8e538b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/241307c3ea1f7c21ecc6edc438291707a8e538b0"}]},"branch":"refs/heads/master"},"cf70ebe98f88ea3ecfda58c17aa1d55c2810d2a7":{"kind":"REWORK","_number":6,"created":"2022-01-12 22:40:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/6"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-12 22:40:09.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/cf70ebe98f88ea3ecfda58c17aa1d55c2810d2a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/cf70ebe98f88ea3ecfda58c17aa1d55c2810d2a7"}]},"branch":"refs/heads/master"},"6271b77c68d19b94d335cbf12b40208cb873f6dd":{"kind":"REWORK","_number":7,"created":"2022-01-12 22:42:27.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/7"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-12 22:42:19.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/6271b77c68d19b94d335cbf12b40208cb873f6dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/6271b77c68d19b94d335cbf12b40208cb873f6dd"}]},"branch":"refs/heads/master"},"538b9de970946e8434c568d06ba32e0d808eb049":{"kind":"REWORK","_number":8,"created":"2022-01-13 10:52:03.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/823710/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/10/823710/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/10/823710/8"}}},"commit":{"parents":[{"commit":"64c82a0829c6c5356ed2a44f14c1ea675aa493b7","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/64c82a0829c6c5356ed2a44f14c1ea675aa493b7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-06 15:48:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-13 10:51:55.000000000","tz":60},"subject":"WIP: Add a new max_sample_default","message":"WIP: Add a new max_sample_default\n\nIn order to address something like this:\nhttps://bugs.launchpad.net/cinder/+bug/1956601\n\nwe need a max_sample_default option. This patch attemps to do it.\n\nChange-Id: I1e4ff0ba57323232736b94e1e49f9e899fc78487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/538b9de970946e8434c568d06ba32e0d808eb049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/538b9de970946e8434c568d06ba32e0d808eb049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
