)]}'
{"id":"openstack%2Foslo.config~855333","triplet_id":"openstack%2Foslo.config~stable%2Fyoga~Idd124082265032c30f83242b612d002d7093aa61","project":"openstack/oslo.config","branch":"stable/yoga","topic":"fix-cliff-issue-stable/yoga","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-09-06 15:11:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd124082265032c30f83242b612d002d7093aa61","subject":"Disable a config test if ran under the root user","status":"MERGED","created":"2022-08-31 08:52:27.000000000","updated":"2022-09-06 15:13:09.000000000","submitted":"2022-09-06 15:11:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855333-fix-cliff-issue-stable/yoga","cherry_pick_of_change":851836,"cherry_pick_of_patch_set":6,"meta_rev_id":"262aad627f0e5c979bfcfafbf7a3780da3132f00","_number":855333,"virtual_id_number":855333,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 15:11:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-09-06 13:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-09-06 13:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 10:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 13:34:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"337b1bfc858d304d7080d4351011cf1fe2b492f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-31 08:52:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed1bb9af88e83b9957380e081906f1eac4c80e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c562aec532684f02900e85392f999c9f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c84eac3e394470bb992be7048cc9834 : SUCCESS in 2h 02m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a236c2c4285f40338205be75a0146db2 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8856d45ab1f64f648e46442998c6e564 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/968c468b821d4722b09d76fe7ab22b39 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a154af9a4fb4ce99a90e9c8e93a7775 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"694897254538ddbb8dcf496438fa717ef7f40f17","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 13:34:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e007ed770e71e32debc86630008fd70cc979b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:35:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13f69b8539d989c3ab45728999b826dadf7f8513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:11:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edf9d23074f04d1dba13bcdf001798ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6880c28fef1740ea8268e0aecae12631 : SUCCESS in 1h 35m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c9c451888c40d5a851862a47b45237 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a8d74ef54714e5081481cb21fcebde5 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2cdc8d2bcfa4b998e106b001540b51d : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c4c2b94be04470a9d73a5495b4a04d : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a71b22117ff784d6503fb48a26370c596b04e597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:11:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"262aad627f0e5c979bfcfafbf7a3780da3132f00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:13:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53fcce951f547c4b481600d778fe380\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b100f9454a6643fe991f6b5fedb9e058 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"966c588a0c5b973a324ebe1390f8a0f064a44540","revisions":{"966c588a0c5b973a324ebe1390f8a0f064a44540":{"kind":"REWORK","_number":1,"created":"2022-08-31 08:52:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/855333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/33/855333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/33/855333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/33/855333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/33/855333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/33/855333/1"}}},"commit":{"parents":[{"commit":"e63b6729d0942bdd96b68735b10efd8368e9568d","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/e63b6729d0942bdd96b68735b10efd8368e9568d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-08-02 10:27:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-08-31 08:52:27.000000000","tz":0},"subject":"Disable a config test if ran under the root user","message":"Disable a config test if ran under the root user\n\nThe `test_conf_file_permission_denied` is designed to test\nif a config file can\u0027t be read due to missing file permissions.\n\nHowever if we run unit tests with the root user we will beneficiat\nfrom the root privileges and so this test will fails. This can\nbe reproduced by running `tox` with `sudo`.\n\nThese changes propose to skip this test if root privileges are\ndetected.\n\nCloses-bug: #1983358\n\nChange-Id: Idd124082265032c30f83242b612d002d7093aa61\n(cherry picked from commit 7b838f5c31c3219d13661ec612d750e58683cbe6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/966c588a0c5b973a324ebe1390f8a0f064a44540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/966c588a0c5b973a324ebe1390f8a0f064a44540"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
