)]}'
{"id":"openstack%2Foslo.config~874209","triplet_id":"openstack%2Foslo.config~master~I7be9a22796f484df79f0f509a2cf4f47e3a00128","project":"openstack/oslo.config","branch":"master","topic":"release-health-check-oslo","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-21 13:54:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I7be9a22796f484df79f0f509a2cf4f47e3a00128","subject":"Fix issues related to tox4","status":"MERGED","created":"2023-02-17 09:22:31.000000000","updated":"2023-02-21 15:15:37.000000000","submitted":"2023-02-21 15:14:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"874209-release-health-check-oslo","meta_rev_id":"b73e20f5955c268a360d46b6f492c4715b98f6bd","_number":874209,"virtual_id_number":874209,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 15:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-02-21 13:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-02-21 13:54:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 10:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 15:32:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"35d1116c4c3f8d456b69ea3516c36052881b9d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"446e9a5efcc56ebee00a85a7be096854b7424e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:34:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cf630291cc54110b818994e4cd6b760\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c76609b8d7494a6995c18dc67b586e95 : FAILURE in 1h 09m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e6ba9146fa4b7c8546f3e99f08eab0 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78ba3af370fc477a9048915a5f02e515 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04dfc7b04cac449fb2a46e0044cd0447 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68040e7aca249f58ddc6b9950ce1b4b : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a29674c04ec84193b82b0e67f9013cf1 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7ca759e74dd89f2af2f55c4cf2f2b44a4630fe7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:07:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5183ba078e96f47f5040755ee3bcde70b2cfc813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:10:40.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5e513291f5df1bb1e207a3c1fd905c07d4b3b3fb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:10:45.000000000","message":"Topic oslo-tox-4 removed","accounts_in_message":[],"_revision_number":3},{"id":"a2ce02978c16a69213c8b063a4be031097181937","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 13:11:01.000000000","message":"Topic set to release-health-check-oslo","accounts_in_message":[],"_revision_number":3},{"id":"237397cf380038fca928c5db4ddf4fe07519cfa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:26:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bbc9324ed33449495ae2441daf1d46b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3100e915433545b68a081cd576cde63a : SUCCESS in 1h 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fbfb8b39b24dc79ea76d5670306703 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a717fa491219475c8ea43ded697ca5ab : POST_FAILURE in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4daf2bc6ca6b494d9221c87cca0fd392 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9c0fe3964c481cbad5c18b5b0c2113 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e7083fa1d93489d9bb5608a8a674444 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f7b962e2ced92b19fc8cc31433895145e0299738","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 14:30:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5adf9bf17654506af2af9281f14cd0494f6aab7d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 15:32:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5e9f6bd05e793b1d6481d406170c3e96255bb5a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:53:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f944c0cf8d424de7974f35dda1a17a8a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa7832a3b2f44a4f9b9270dee64a7dfb : SUCCESS in 1h 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18ad3473a514d7a9c867170bec7eaae : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d8ff9a3a4d48df9d6f43ffb0009c94 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3bf2de75eb24ac1afe0d797ae743402 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e4e91b8bda4013b174f3b8a2562c01 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9721606f611643f5bf3bf75b5a363e92 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"c0f3fd713ba3320616f597fd909746cd4682af97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 16:49:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ea5593c559c061643036368a664b83b2a5166ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:13:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f621b0634a8942bba06c959d1f6885a9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/549d35eb0e664199b49696bb6e1a5a45 : SUCCESS in 1h 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52bb01a81a54401aa06f56b1978e384e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa80ed9d52643189159a859624ba1af : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7a74cb3be1842b79891c202e828bcf4 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd575b53c6e24cd7966e66f5268a2a41 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc3bb70fdefc4f41b19b645db199a46a : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2127bc312b8bdcdad9d81d4d1161bb09adf392af","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 13:54:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"633b6a89b7c4e17c014f9a3e473bfe6ae07f7286","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 13:54:39.000000000","message":"Topic oslo-tox-4 removed","accounts_in_message":[],"_revision_number":4},{"id":"888eac331647f8e29e54b07b1eeb75becbffa07a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 13:54:43.000000000","message":"Topic set to release-health-check-oslo","accounts_in_message":[],"_revision_number":4},{"id":"625b57cf0ddf8e60f4ba5e9763ca2a823d1dd878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:54:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"11c58bd58ce7ae5d74424c43611de4aa55854571","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:14:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d6feb9d24e7402487184d3c6daa9b33\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbe5a3e815a44e968cdfa890452bf2a3 : SUCCESS in 1h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4e4372aefa480981df0b476d628475 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d23a816ccae4af091a626b4d12a2928 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70d521d020b64803b057ced9e89e035f : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50eaf7d41c64195ae7b99508fc2b5ed : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7264b2675154bd195c6c458ad0b3939 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b785c8bfbf9fee7dfdb9250fe95dca124ba994db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:14:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b73e20f5955c268a360d46b6f492c4715b98f6bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:15:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e18d0699084d66aba152814f248914\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47702b5abaff4d97a5a73c9744383a89 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3bb89d6250e49ed90e8e1df93c8320a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"515daab81acccbea4015afc5c3c3316d88903a10","revisions":{"c871f0fdd2e13a7c74dd468d9cd3986093f28714":{"kind":"REWORK","_number":1,"created":"2023-02-17 09:22:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/09/874209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/09/874209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/1"}}},"commit":{"parents":[{"commit":"84478d83f87e9993625044de5cd8b4a18dfcaf5d","subject":"Make tests backward compat with cliff\u0027 older versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/84478d83f87e9993625044de5cd8b4a18dfcaf5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:22:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:22:25.000000000","tz":60},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nChange-Id: I7be9a22796f484df79f0f509a2cf4f47e3a00128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/c871f0fdd2e13a7c74dd468d9cd3986093f28714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/c871f0fdd2e13a7c74dd468d9cd3986093f28714"}]},"branch":"refs/heads/master"},"9cb7ad7c7358adb8687aad7c1c05f03db4653b22":{"kind":"REWORK","_number":2,"created":"2023-02-17 13:07:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/09/874209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/09/874209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/2"}}},"commit":{"parents":[{"commit":"84478d83f87e9993625044de5cd8b4a18dfcaf5d","subject":"Make tests backward compat with cliff\u0027 older versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/84478d83f87e9993625044de5cd8b4a18dfcaf5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:22:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 13:07:02.000000000","tz":60},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nChange-Id: I7be9a22796f484df79f0f509a2cf4f47e3a00128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/9cb7ad7c7358adb8687aad7c1c05f03db4653b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/9cb7ad7c7358adb8687aad7c1c05f03db4653b22"}]},"branch":"refs/heads/master"},"77f8b362955d489a37be2e04117ff17341fe7723":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-17 13:10:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/09/874209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/09/874209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/3"}}},"commit":{"parents":[{"commit":"84478d83f87e9993625044de5cd8b4a18dfcaf5d","subject":"Make tests backward compat with cliff\u0027 older versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/84478d83f87e9993625044de5cd8b4a18dfcaf5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:22:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-17 13:10:40.000000000","tz":0},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nRelated to https://lists.openstack.org/pipermail/openstack-discuss/2023-February/032247.html\n\nChange-Id: I7be9a22796f484df79f0f509a2cf4f47e3a00128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/77f8b362955d489a37be2e04117ff17341fe7723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/77f8b362955d489a37be2e04117ff17341fe7723"}]},"branch":"refs/heads/master","description":"Edit commit message"},"515daab81acccbea4015afc5c3c3316d88903a10":{"kind":"REWORK","_number":4,"created":"2023-02-17 16:49:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/09/874209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/09/874209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/09/874209/4"}}},"commit":{"parents":[{"commit":"84478d83f87e9993625044de5cd8b4a18dfcaf5d","subject":"Make tests backward compat with cliff\u0027 older versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/84478d83f87e9993625044de5cd8b4a18dfcaf5d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 09:22:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-17 16:49:38.000000000","tz":60},"subject":"Fix issues related to tox4","message":"Fix issues related to tox4\n\nChange-Id: I7be9a22796f484df79f0f509a2cf4f47e3a00128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/515daab81acccbea4015afc5c3c3316d88903a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/515daab81acccbea4015afc5c3c3316d88903a10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
